)]}'
{"id":"openstack%2Freleases~873533","triplet_id":"openstack%2Freleases~master~I3f0857e681efb816ac871ede4e1b2608e629cad0","project":"openstack/releases","branch":"master","topic":"antelope-milestone-3","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-02-17 12:39:37.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-02-17 12:39:37.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-17 11:01:45.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I3f0857e681efb816ac871ede4e1b2608e629cad0","subject":"Release final python-manilaclient for Antelope","status":"MERGED","created":"2023-02-13 10:08:13.000000000","updated":"2023-02-17 12:39:37.000000000","submitted":"2023-02-17 12:39:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873533-antelope-milestone-3","meta_rev_id":"3d09e1156389f5137636ddc40a71911e944c9ca3","_number":873533,"virtual_id_number":873533,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 12:39:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-02-17 12:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-02-17 11:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-02-17 12:24:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-02-17 10:49:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 10:09:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 10:59:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 12:20:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-02-13 12:20:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-02-17 12:24:42.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"77926748667ef82f26fb962acd90865f3c754dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 10:08:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"359b0b45513182632bae727131e28316d6f5f1cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 10:59:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9662e0d3fd154eafa49364b51214fb5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb367337dbc48ebaeb6f039d8a5abfa : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/923a55cce764429b988abbd50cb470cf : SUCCESS in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9f3b2e87949047f8979d76b632163bcf : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f4ddaf386461fe01f5c2e83111e1a8d01f892be3","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-13 12:25:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ecdef149ef3faa8a6dd745c8a76708de7e0e929c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 13:43:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7fae7e50edcbb46229deca13d39b396bfee9c737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:57:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ea1ba0d68a84a0d9493d18a5eb004cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718e31c006754daf855c4011461eb696 : SUCCESS in 9m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66089f0457b04399aa653cf37ffe05be : SUCCESS in 5m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/76b6fe3a8fd644b490e5dc154f2f0f00 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"0542ec641db84a3629bddb1c9cb60d6e38a8fafc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 14:05:46.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"30f1051f7b5c8a16804eeec5ea0f254ad74e34e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 14:17:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dabf37143335441ea8388f369b79b6e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8549961754404b67a45d253f44e51f61 : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/feb702a8a4aa426a9600600c91457bec : SUCCESS in 6m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c1190ca30591497d8428859b67b15694 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"e7b2626f40d1204e8470608f57062fee4e2a4f66","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-14 14:28:22.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"071e2ceded0abd180d5df83d8a993c85b6990bbd","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-16 13:39:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79ff9ace6ba662fdd9ce273b44a38501e27ab3d3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 09:59:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebb0117f48d095e3a3ff24930e21797c34a15d16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-17 10:48:19.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d9c3ce41f056b26a8712e8286c6ed6b36067af88","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-17 10:48:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2dd448af7c63bd50cedac69c29d374076090aaea","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:49:08.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"df8e5a180b0f8f6bd8e2ddd8093cb29229ca607a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:58:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/983375bd75074cf88651f30cd3283bae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003192ee05c145ff8f4a2e6c55c2628a : SUCCESS in 7m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/011e62db68814b8fafe589944e9d9f20 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f4de60f4591b4760bc8c63efd1d8c0a0 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":4},{"id":"b4153bff4da2e8aa9ea40b29914dd3183444792d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 11:01:45.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b2429eb5621bf91016456705afdfa23c3ccd0240","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-17 12:24:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"daee0f68797069835bd066d1cdc0df1e091ce5b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:25:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8a704d5fe8973b96f4b13bf748844c6f1c8df7be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:39:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/727dc3c6998e405aa878b1bca64946ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d52208c4904d9db20b874ebc8a8731 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cbfabfdbc37405db36f4a88cc2006bb : SUCCESS in 7m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf3654a8e438463e89319982eb932989 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":4},{"id":"3d09e1156389f5137636ddc40a71911e944c9ca3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:39:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ddafa74804e131f0e81ee6a3cc0a817ff7b77f74","revisions":{"f0906907a14077217cc66964117753e413539fc4":{"kind":"REWORK","_number":1,"created":"2023-02-13 10:08:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/873533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/33/873533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/33/873533/1"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:08:09.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:08:09.000000000","tz":60},"subject":"Release final python-manilaclient for Zed","message":"Release final python-manilaclient for Zed\n\nThis is a release for python-manilaclient to get any merged but\nunreleased changes out prior to the client lib freeze for Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 16th.\n\n$ git log --oneline --no-merges 4.2.0..8399923\nbbfd7d5 Metadata for Share Snapshots\n782e85d Drop duplicate tox config\nbdd268e Fix tox4 errors\n43943fd Support --os-key option\ncb14e96 Fix functional tests failing with python 3.10\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I3f0857e681efb816ac871ede4e1b2608e629cad0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0906907a14077217cc66964117753e413539fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0906907a14077217cc66964117753e413539fc4"}]},"branch":"refs/heads/master"},"dcd87887ffb5df97b0e74003af7f9d6382a2ddd3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-13 13:43:48.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/873533/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/33/873533/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/33/873533/2"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:08:09.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-13 13:43:48.000000000","tz":0},"subject":"Release final python-manilaclient for Zed","message":"Release final python-manilaclient for Zed\n\nThis is a release for python-manilaclient to get any merged but\nunreleased changes out prior to the client lib freeze for Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 4.2.0..8399923\nbbfd7d5 Metadata for Share Snapshots\n782e85d Drop duplicate tox config\nbdd268e Fix tox4 errors\n43943fd Support --os-key option\ncb14e96 Fix functional tests failing with python 3.10\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I3f0857e681efb816ac871ede4e1b2608e629cad0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dcd87887ffb5df97b0e74003af7f9d6382a2ddd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dcd87887ffb5df97b0e74003af7f9d6382a2ddd3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bcbd320047698cc705945e43bc6f4ba6a3a48480":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-13 14:05:46.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/873533/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/33/873533/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/33/873533/3"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:08:09.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-13 14:05:46.000000000","tz":0},"subject":"Release final python-manilaclient for Antelope","message":"Release final python-manilaclient for Antelope\n\nThis is a release for python-manilaclient to get any merged but\nunreleased changes out prior to the client lib freeze for Antelope going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 4.2.0..8399923\nbbfd7d5 Metadata for Share Snapshots\n782e85d Drop duplicate tox config\nbdd268e Fix tox4 errors\n43943fd Support --os-key option\ncb14e96 Fix functional tests failing with python 3.10\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I3f0857e681efb816ac871ede4e1b2608e629cad0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bcbd320047698cc705945e43bc6f4ba6a3a48480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bcbd320047698cc705945e43bc6f4ba6a3a48480"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ddafa74804e131f0e81ee6a3cc0a817ff7b77f74":{"kind":"REWORK","_number":4,"created":"2023-02-17 10:48:19.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/33/873533/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/33/873533/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/873533/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/33/873533/4"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:08:09.000000000","tz":60},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-02-17 10:47:56.000000000","tz":0},"subject":"Release final python-manilaclient for Antelope","message":"Release final python-manilaclient for Antelope\n\nThis is a release for python-manilaclient to get any merged but\nunreleased changes out prior to the client lib freeze for Antelope going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 4.2.0..0748620\na0f6a2e Add quiesce_wait_time for replica promote\nbbfd7d5 Metadata for Share Snapshots\n782e85d Drop duplicate tox config\nbdd268e Fix tox4 errors\n43943fd Support --os-key option\ncb14e96 Fix functional tests failing with python 3.10\n05135dc Fix formatting of OSC server migration check-only\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I3f0857e681efb816ac871ede4e1b2608e629cad0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ddafa74804e131f0e81ee6a3cc0a817ff7b77f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ddafa74804e131f0e81ee6a3cc0a817ff7b77f74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
