)]}'
{"id":"openstack%2Freleases~873540","triplet_id":"openstack%2Freleases~master~I6fcad5d8c1cf5adde0c640014665658a86ee3df0","project":"openstack/releases","branch":"master","topic":"antelope-milestone-3","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-02-24 09:08:00.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-21 11:05:58.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"14107":{"account":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"last_update":"2023-02-20 11:45:11.000000000","reason":"\u003cGERRIT_ACCOUNT_14107\u003e replied on the change","reason_account":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}},"hashtags":[],"change_id":"I6fcad5d8c1cf5adde0c640014665658a86ee3df0","subject":"Release final python-solumclient for Antelope","status":"MERGED","created":"2023-02-13 10:13:02.000000000","updated":"2023-02-24 09:24:44.000000000","submitted":"2023-02-24 09:24:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873540-antelope-milestone-3","meta_rev_id":"4bd2aabeb1d3516987d982ab100315e5549f853c","_number":873540,"virtual_id_number":873540,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-24 09:24:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-02-24 09:08:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-02-21 11:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-02-24 09:08:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 10:13:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 10:48:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 12:21:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2023-02-17 11:51:23.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"89506a864ec2c774ac03e89a8b6ad80e5c28e003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 10:13:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b24aaea11b2e18526e575a74fdbf606775e3e616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 10:48:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ded365bce707483093dc0dc20079d6f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9598b56bf01c4ac69170d88082951e35 : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e4f66984ec942f5b624a019244b3978 : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cab01d6038124c08909fc9eb9addea6d : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4358f9fc86bd1a340c844a24d74dcc3864493a2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 13:46:54.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b762345154fedd7e42826a63637e4016e32c26df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:59:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e17ed460696d430cb493caebc52cdab7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae850c69cef433f9dc4d0b0a5f7a057 : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/937053ca2e0147d7baf997ea1882cc28 : SUCCESS in 5m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/200dca109726405a90453267549a895a : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"dad1e1e38db1bcce540452014d5cbb1559fbc38d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 14:06:29.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"77606e3c7eff167f5b7f069e4f27c4cb71457b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 14:29:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09362f468aac44788cd4fc70c392e9b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f42532c7166d4121a9620c8006264d9f : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f2fb32ad78a4c1787844af372fe7958 : SUCCESS in 5m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d47a7a0385fa46ce8fadb05406e9affc : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":3},{"id":"b1d9315ec5b8a5a51fd8be6604b46145ad393285","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 10:13:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"471c9b3d6f3bf2dfda5f97d7245e055cd3d89b8e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-17 11:51:23.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a7def0043c8483acbae3015d51a2fcde563734d","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2023-02-20 11:45:11.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b86b1c57bc1976547a31967479c1b04166a58423","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 11:46:04.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"fb042bf86966aa700ed5189235d2b51772e2207d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-20 13:30:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"353d6c8c96e23950ec00e7f453cb40b96e62090f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 13:38:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ece20bb98804567aa05eb314eede252\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e24081684a14683bd71f58be28b7f1b : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1df95d4e1d824cfda434fb42d7327199 : SUCCESS in 5m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c7a74a7cbfed46c785fe8bbee151746c : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":4},{"id":"d736ef3a8afb4a9edfabe29da06329858d7c2db8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 11:05:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"97f8c71ea8fde12f467ec26baa0cf9596fa72c89","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-24 09:08:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"54c9cf78faa8ade15b37c688412579b7022291b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 09:08:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e78b308f5c490e2ba7ed303c1b66940e18e2f578","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 09:24:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5bf44cf3f8a49cfbc40d725bf59aa2e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c23d09ad84ab43e99a766f92d3151efc : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb812084193649c9bfba32077598c389 : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31de0992c67d434a99e763b1f4b26d32 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":4},{"id":"4bd2aabeb1d3516987d982ab100315e5549f853c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 09:24:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"431a92c28aa9842d810faf5f6249ebe0ed98e91e","revisions":{"887bb8b59706f8e0dc29819ca84b01fbbf202fe7":{"kind":"REWORK","_number":1,"created":"2023-02-13 10:13:02.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/873540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/873540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/873540/1"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:12:58.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:12:58.000000000","tz":60},"subject":"Release final python-solumclient for Zed","message":"Release final python-solumclient for Zed\n\nThis is a release for python-solumclient to get any merged but\nunreleased changes out prior to the client lib freeze for Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 16th.\n\n$ git log --oneline --no-merges 3.6.1..2afd925\n2afd925 Switch to 2023.1 Python3 unit tests and generic template name\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I6fcad5d8c1cf5adde0c640014665658a86ee3df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/887bb8b59706f8e0dc29819ca84b01fbbf202fe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/887bb8b59706f8e0dc29819ca84b01fbbf202fe7"}]},"branch":"refs/heads/master"},"dcfa83f2e4ba9ee72047e078273203ed680897b4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-13 13:46:54.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/873540/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/873540/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/873540/2"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:12:58.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-13 13:46:54.000000000","tz":0},"subject":"Release final python-solumclient for Zed","message":"Release final python-solumclient for Zed\n\nThis is a release for python-solumclient to get any merged but\nunreleased changes out prior to the client lib freeze for Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 3.6.1..2afd925\n2afd925 Switch to 2023.1 Python3 unit tests and generic template name\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I6fcad5d8c1cf5adde0c640014665658a86ee3df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dcfa83f2e4ba9ee72047e078273203ed680897b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dcfa83f2e4ba9ee72047e078273203ed680897b4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9431cc6b6ad605953e46f45a1132b27cf8ea07a4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-13 14:06:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/873540/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/873540/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/873540/3"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:12:58.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-13 14:06:29.000000000","tz":0},"subject":"Release final python-solumclient for Antelope","message":"Release final python-solumclient for Antelope\n\nThis is a release for python-solumclient to get any merged but\nunreleased changes out prior to the client lib freeze for Antelope going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 3.6.1..2afd925\n2afd925 Switch to 2023.1 Python3 unit tests and generic template name\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I6fcad5d8c1cf5adde0c640014665658a86ee3df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9431cc6b6ad605953e46f45a1132b27cf8ea07a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9431cc6b6ad605953e46f45a1132b27cf8ea07a4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"431a92c28aa9842d810faf5f6249ebe0ed98e91e":{"kind":"REWORK","_number":4,"created":"2023-02-20 13:30:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/40/873540/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/873540/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/873540/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/873540/4"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:12:58.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-02-20 13:29:12.000000000","tz":60},"subject":"Release final python-solumclient for Antelope","message":"Release final python-solumclient for Antelope\n\nThis is a release for python-solumclient to get any merged but\nunreleased changes out prior to the client lib freeze for Antelope going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 3.6.1..b9b0d83\nb9b0d83 Fix tox 4 compatibility\n2afd925 Switch to 2023.1 Python3 unit tests and generic template name\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I6fcad5d8c1cf5adde0c640014665658a86ee3df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/431a92c28aa9842d810faf5f6249ebe0ed98e91e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/431a92c28aa9842d810faf5f6249ebe0ed98e91e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
