)]}'
{"id":"openstack%2Freleases~873548","triplet_id":"openstack%2Freleases~master~I2eed7365e759e94a08c14df29d7ade98b968bf9d","project":"openstack/releases","branch":"master","topic":"antelope-milestone-3","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2023-02-13 12:40:39.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-17 16:01:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2eed7365e759e94a08c14df29d7ade98b968bf9d","subject":"Release final python-zunclient for Antelope","status":"MERGED","created":"2023-02-13 10:16:31.000000000","updated":"2023-02-17 16:01:42.000000000","submitted":"2023-02-17 16:01:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873548-antelope-milestone-3","meta_rev_id":"bebe3bb6c05394da879ec26bd92c7219d6159f38","_number":873548,"virtual_id_number":873548,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 16:01:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2023-02-17 15:30:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-02-17 14:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2023-02-17 15:30:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 10:17:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 11:04:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 12:20:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2023-02-17 15:30:24.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"0881588c94a64f69df786ee823560a183419c0d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 10:16:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e11528c3d9d1f5a95a8a1e034100cb44e79a9c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:04:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb1674358a6f430fa424aa6d67690198\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3538a3d2a07423ca0a0c4c17a100b21 : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e27a3ba7df5b4972a60267e52496330e : SUCCESS in 7m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4778c87d5fef430b9a47f77d8cb3cb5b : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e7f9485626d4265f16c3119f9e453fa612c7af5d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-02-13 12:40:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"74754af260634c47ea5f8bec0d8b7e2d822b3aca","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 12:41:28.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"de24d7c5ca4ae0d7a517808ec85eaaf463a7b80d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 13:43:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"eebbdb55cfd1d549e41039cf46e31a45185bc8d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:52:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dcb7b43c1604788877254210069910f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/807de5fca8854aef8b594d0139c0924f : SUCCESS in 7m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/adfb5b60d64d43e9956be1d1a0874031 : SUCCESS in 6m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2a18ad74ee4a4192965c84b5e2602632 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"72a4bfd74af4e751cf1617f6236c98c539930278","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 13:58:09.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4f94b7f83d06b1d6d744e4088176651de5b7439c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 14:06:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b25c1c5dd3a4a108651b2ad9cbb2024\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf88fc7f93442f6b42b9e0826eb0a8e : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f1dd5a526505466b8d32027215fd6a79 : SUCCESS in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ef6296b1d20448308374752e63f48a54 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"f1c9d387bb15da0af34b0792dc092237a7d63fb5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 10:14:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7df7c3633a53f6f8bbcb8ce8f4254c9040cf11ce","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 14:07:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b7730a734bd661308d81872a7d4cfbe7fd3ed601","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-17 15:30:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6456a54deec110bbdad949b891982430a44bf154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:30:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7196b3c8f7395614040d74d985ea388136dcba4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:01:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b00b5bc5c464404ab85ca585d127e65\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1807c9de5e2a4d6a87706dd37b7f9102 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/284e244a8adb47378b7c69a0a12dfc74 : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/313744b03435419d8290b7b001baeb71 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":3},{"id":"bebe3bb6c05394da879ec26bd92c7219d6159f38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:01:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4ad46dbeb9e04ee840595e636b9edba253c0bb37","revisions":{"543b191b4393aac9b90b0770e15895bf54a12f0a":{"kind":"REWORK","_number":1,"created":"2023-02-13 10:16:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/48/873548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/873548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/873548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/873548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/873548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/873548/1"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:16:27.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:16:27.000000000","tz":60},"subject":"Release final python-zunclient for Zed","message":"Release final python-zunclient for Zed\n\nThis is a release for python-zunclient to get any merged but\nunreleased changes out prior to the client lib freeze for Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 16th.\n\n$ git log --oneline --no-merges 4.5.0..e586422\ne586422 Fix tox4 error\n791efa6 Switch to 2023.1 Python3 unit tests and generic template name\n01d5257 Update master for stable/zed\nd38f743 Add Python3 zed unit tests\nab5b700 Drop lower-constraints.txt and its testing\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I2eed7365e759e94a08c14df29d7ade98b968bf9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/543b191b4393aac9b90b0770e15895bf54a12f0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/543b191b4393aac9b90b0770e15895bf54a12f0a"}]},"branch":"refs/heads/master"},"0376f23e7fc02c136c91e9a5adb37b48e784c36c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-13 13:43:01.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/48/873548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/873548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/873548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/873548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/873548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/873548/2"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:16:27.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-13 13:43:01.000000000","tz":0},"subject":"Release final python-zunclient for Zed","message":"Release final python-zunclient for Zed\n\nThis is a release for python-zunclient to get any merged but\nunreleased changes out prior to the client lib freeze for Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 4.5.0..e586422\ne586422 Fix tox4 error\n791efa6 Switch to 2023.1 Python3 unit tests and generic template name\n01d5257 Update master for stable/zed\nd38f743 Add Python3 zed unit tests\nab5b700 Drop lower-constraints.txt and its testing\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I2eed7365e759e94a08c14df29d7ade98b968bf9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0376f23e7fc02c136c91e9a5adb37b48e784c36c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0376f23e7fc02c136c91e9a5adb37b48e784c36c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4ad46dbeb9e04ee840595e636b9edba253c0bb37":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-13 13:58:09.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/48/873548/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/873548/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/873548/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/873548/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/873548/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/873548/3"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:16:27.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-13 13:58:09.000000000","tz":0},"subject":"Release final python-zunclient for Antelope","message":"Release final python-zunclient for Antelope\n\nThis is a release for python-zunclient to get any merged but\nunreleased changes out prior to the client lib freeze for Antelope going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 4.5.0..e586422\ne586422 Fix tox4 error\n791efa6 Switch to 2023.1 Python3 unit tests and generic template name\n01d5257 Update master for stable/zed\nd38f743 Add Python3 zed unit tests\nab5b700 Drop lower-constraints.txt and its testing\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I2eed7365e759e94a08c14df29d7ade98b968bf9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ad46dbeb9e04ee840595e636b9edba253c0bb37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ad46dbeb9e04ee840595e636b9edba253c0bb37"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
