)]}'
{"id":"openstack%2Freleases~874543","triplet_id":"openstack%2Freleases~master~I9eac56962a8247837a37aff1d531ba0ce9503e53","project":"openstack/releases","branch":"master","topic":"nova_antelope_highlights","attention_set":{},"removed_from_attention_set":{"16708":{"account":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"last_update":"2023-02-24 17:27:47.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-02-24 17:27:47.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-02-24 17:27:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9eac56962a8247837a37aff1d531ba0ce9503e53","subject":"nova c-h: fup for adding link for the cpu management docs","status":"MERGED","created":"2023-02-21 09:38:09.000000000","updated":"2023-02-24 17:27:47.000000000","submitted":"2023-02-24 17:27:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874543-nova_antelope_highlights","meta_rev_id":"c2fcf1f5a28fc79511f315c130b3568361e264ed","_number":874543,"virtual_id_number":874543,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-24 17:27:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-02-24 17:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-02-23 17:00:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-02-24 17:03:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-02-22 11:06:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 09:39:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 12:21:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-02-23 17:00:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-02-23 17:55:23.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2023-02-24 17:03:25.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"d291a86ca0485688cabafbdd353b962b73a17eb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-02-21 09:38:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27c1454a4676009d5f075c005049855d9ff1e2b2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 09:39:07.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"4d4b0a78f1777d72beeb9b42103c5792d4490e33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 09:46:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f9e3b7d37e48dea7f77803b1640fcb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad30ec9aa90c48388022f389b6cacaa5 : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81a1b6753adf4e6eafe38464101fe5ea : SUCCESS in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8509fba5ef714239a326fb4f98610548 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ec740d50b98bdc443d0c499983e0291d09ec5ff9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-02-22 10:07:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"82df256b269d846af9c8fab60bcb08f6809e1e9e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:09:30.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"3ce608d2b5d5dfb44a51fb25a6d357ac0c5c1d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:18:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6ea71efdf3f4788b4e40353669f12e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e7310806a843ba8f716cbe412b0702 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96eb4864e6d54721b4b550dd1018b470 : SUCCESS in 5m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0ac9bf4eca564848a1874069472f43f9 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"9bbcb4f7a91004364fa0fffe2e64a7157648077c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-02-22 11:03:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e946ea6374046946fae3272faa41b3cf45cee23e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 11:04:51.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"c6ac48b097c64f1fd91a1bc98574d3edec5fe30c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-02-22 11:05:17.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"1f0ca006b3cc4d4a218051c619807f68c4ed9a39","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 11:06:29.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"4d2621534a169ecd63181dfeff859aeb8b9ec620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 11:18:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da54a1c07094beeb20afa7012dc65d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ccb3484949f4d7aab90dcf5c7df298d : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9ec4656642f41f09c4c20882e5f94c7 : SUCCESS in 6m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f557a96e983a418d8b5bdb75fd4a9468 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":4},{"id":"fc6f3b70f7ef6fa30cfb73837df5908ba106e6a4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-23 17:00:55.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cb453c45380263affd4115d7c1a3e8eaeaa3614b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-02-24 17:03:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bb1fdd246f2ace1bf8241939b3441014c4020496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:07:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fdcfc228b33b222200e6aa36e3736754b92a7ff0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:27:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da43fedbbc864299bfb763d5339d35d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e653161a7dda41bbb1a4a5b4b44ee54e : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d8865695334a4594cc82fe6f01bdd6 : SUCCESS in 7m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dea63e3d83e94d49acf0ab1a9dea1cef : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":4},{"id":"c2fcf1f5a28fc79511f315c130b3568361e264ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:27:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c03e3edf4725e54e8a35481f303c0cc1309d28af","revisions":{"9fed376bbe42fc4000b0cd88935acda54c47cd3f":{"kind":"REWORK","_number":1,"created":"2023-02-21 09:38:09.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/43/874543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/874543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/874543/1"}}},"commit":{"parents":[{"commit":"4b8fa804df06a2e9b53c731e3429fff4241bc41f","subject":"Nova 2023.1 Antelope cycle highlights","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b8fa804df06a2e9b53c731e3429fff4241bc41f"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-02-21 09:36:12.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-02-21 09:36:12.000000000","tz":60},"subject":"nova c-h: fup for adding link for the cpu management docs","message":"nova c-h: fup for adding link for the cpu management docs\n\nI just created a doc patch for how to use cpu management in nova.\nThat doc would be nice for the related cycle highlights.\n\nChange-Id: I9eac56962a8247837a37aff1d531ba0ce9503e53\nDepends-On: https://review.opendev.org/c/openstack/nova/+/874514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9fed376bbe42fc4000b0cd88935acda54c47cd3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9fed376bbe42fc4000b0cd88935acda54c47cd3f"}]},"branch":"refs/heads/master"},"170fcf7dff953ffb04e06127e6432da5368d295e":{"kind":"REWORK","_number":2,"created":"2023-02-22 10:07:12.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/43/874543/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/874543/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/874543/2"}}},"commit":{"parents":[{"commit":"558ddcb1368bd5b744ea412503a8a36db5af0874","subject":"Nova 2023.1 Antelope cycle highlights","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/558ddcb1368bd5b744ea412503a8a36db5af0874"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-02-21 09:36:12.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-02-22 10:07:03.000000000","tz":60},"subject":"nova c-h: fup for adding link for the cpu management docs","message":"nova c-h: fup for adding link for the cpu management docs\n\nI just created a doc patch for how to use cpu management in nova.\nThat doc would be nice for the related cycle highlights.\n\nChange-Id: I9eac56962a8247837a37aff1d531ba0ce9503e53\nDepends-On: https://review.opendev.org/c/openstack/nova/+/874514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/170fcf7dff953ffb04e06127e6432da5368d295e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/170fcf7dff953ffb04e06127e6432da5368d295e"}]},"branch":"refs/heads/master"},"6bd886f70f7445a1825cc3cac2319ffee08f5f9f":{"kind":"REWORK","_number":3,"created":"2023-02-22 11:03:43.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/43/874543/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/874543/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/874543/3"}}},"commit":{"parents":[{"commit":"f10c9b5c3a6ee5c0fe173a14f7ea021065a4de67","subject":"Nova 2023.1 Antelope cycle highlights","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f10c9b5c3a6ee5c0fe173a14f7ea021065a4de67"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-02-21 09:36:12.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-02-22 11:03:34.000000000","tz":60},"subject":"nova c-h: fup for adding link for the cpu management docs","message":"nova c-h: fup for adding link for the cpu management docs\n\nI just created a doc patch for how to use cpu management in nova.\nThat doc would be nice for the related cycle highlights.\n\nChange-Id: I9eac56962a8247837a37aff1d531ba0ce9503e53\nDepends-On: https://review.opendev.org/c/openstack/nova/+/874514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bd886f70f7445a1825cc3cac2319ffee08f5f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bd886f70f7445a1825cc3cac2319ffee08f5f9f"}]},"branch":"refs/heads/master"},"c03e3edf4725e54e8a35481f303c0cc1309d28af":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-22 11:05:17.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/43/874543/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/874543/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/874543/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/874543/4"}}},"commit":{"parents":[{"commit":"ec0b4ec312aef446827b545bf4f6bed625c1d82f","subject":"Nova 2023.1 Antelope cycle highlights","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ec0b4ec312aef446827b545bf4f6bed625c1d82f"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-02-21 09:36:12.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-02-22 11:05:09.000000000","tz":60},"subject":"nova c-h: fup for adding link for the cpu management docs","message":"nova c-h: fup for adding link for the cpu management docs\n\nI just created a doc patch for how to use cpu management in nova.\nThat doc would be nice for the related cycle highlights.\n\nChange-Id: I9eac56962a8247837a37aff1d531ba0ce9503e53\nDepends-On: https://review.opendev.org/c/openstack/nova/+/874514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c03e3edf4725e54e8a35481f303c0cc1309d28af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c03e3edf4725e54e8a35481f303c0cc1309d28af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
