)]}'
{"id":"openstack%2Freleases~875389","triplet_id":"openstack%2Freleases~master~I0f85d10f1ca56dc643f3cfcc0081665e65c8f234","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-03-01 13:36:19.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-03-01 13:36:19.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-01 13:06:28.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2023-03-01 12:44:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9914\u003e replied on the change","reason_account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}},"hashtags":[],"change_id":"I0f85d10f1ca56dc643f3cfcc0081665e65c8f234","subject":"Release ansible-role-thales-hsm RC1 for Antelope","status":"MERGED","created":"2023-02-27 11:01:19.000000000","updated":"2023-03-01 13:36:19.000000000","submitted":"2023-03-01 13:36:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875389-antelope-rc1-deadline","meta_rev_id":"aaca526aaa361a11b346b792dcdb51d1dacfafc8","_number":875389,"virtual_id_number":875389,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-01 13:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-03-01 12:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-03-01 12:44:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":2,"date":"2023-03-01 13:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":1,"date":"2023-03-01 13:06:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-03-01 12:45:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 11:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 11:14:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 12:24:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2023-02-27 12:24:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2023-02-27 12:24:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2023-03-01 12:59:41.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"3c2564acac611ee94853f1c40ccb07a72ca0b75b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 11:01:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b4b826954470f2a473deb7403432b6d3bf75ef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 11:14:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb4fa4f0f68d46b8b2ec2b6d1d1af545\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8563c829884f468bba244b1bb9a08824 : SUCCESS in 7m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/257bd5182e5f4d1d8b665059fd7c442e : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7360c767480447aaae51f3e04d9f2074 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"404f3feb545316221041b259c7a99bc8441e122b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:26:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3ccb12ec892d81ff6c9a7335a185bbc6b1cd6001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:35:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58877234ed9248b6b721f5eaf380f23e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9369abf1fe424a3f8ab824db457e9655 : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b296fce8b7fe440e9eb4b56283180e41 : SUCCESS in 5m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/960042bd89e14359995bde3d0e9174c9 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"2c2a46d1d8560dfd2d9944371c74f9c487f1872b","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-03-01 12:44:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0e51addc299caa2742e6b066a280e2baf34249ef","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 12:45:15.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"50bb3f50f08e8f0eec8be5568fc26a93a3d3d46c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-01 12:59:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b4266fb8ecc9b9068e47741aebd4f0b60067ceb0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-01 13:06:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"27b6090cceba40d61bd4d537bf6a1e4264d72fa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 13:07:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b8042cd54a7496de5ec97cce18fe720e1df3281f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 13:36:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92dfd3d38db5480691e1352fb6ec8156\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2c1657f13a48ee87f199d212884057 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/603ebd8514f54a07ba6679bc1a3276b7 : SUCCESS in 8m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48ab2cd16e594ac9a4732cc7c3cd2537 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":2},{"id":"aaca526aaa361a11b346b792dcdb51d1dacfafc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 13:36:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ebf37077d198fa6491600f4ff0614941e2c0b8d","revisions":{"08767cf6c959c1664ac75aa155239a11aed46671":{"kind":"REWORK","_number":1,"created":"2023-02-27 11:01:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/875389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/875389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/875389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/875389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/875389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/875389/1"}}},"commit":{"parents":[{"commit":"bda9d1613b39b8521d5e2f66b780080ae47d845a","subject":"Merge \"Release networking-baremetal 6.1.0 for antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bda9d1613b39b8521d5e2f66b780080ae47d845a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:01:15.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:01:15.000000000","tz":60},"subject":"Release ansible-role-thales-hsm RC1 for Antelope","message":"Release ansible-role-thales-hsm RC1 for Antelope\n\nThis proposed a RC1 release for ansible-role-thales-hsm to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 4.0.0..49970b3\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I0f85d10f1ca56dc643f3cfcc0081665e65c8f234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/08767cf6c959c1664ac75aa155239a11aed46671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/08767cf6c959c1664ac75aa155239a11aed46671"}]},"branch":"refs/heads/master"},"2ebf37077d198fa6491600f4ff0614941e2c0b8d":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:26:30.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/875389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/875389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/875389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/875389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/875389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/875389/2"}}},"commit":{"parents":[{"commit":"bda9d1613b39b8521d5e2f66b780080ae47d845a","subject":"Merge \"Release networking-baremetal 6.1.0 for antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bda9d1613b39b8521d5e2f66b780080ae47d845a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:01:15.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 14:26:30.000000000","tz":0},"subject":"Release ansible-role-thales-hsm RC1 for Antelope","message":"Release ansible-role-thales-hsm RC1 for Antelope\n\nThis proposed a RC1 release for ansible-role-thales-hsm to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 4.0.0..49970b3\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I0f85d10f1ca56dc643f3cfcc0081665e65c8f234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ebf37077d198fa6491600f4ff0614941e2c0b8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ebf37077d198fa6491600f4ff0614941e2c0b8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
