)]}'
{"id":"openstack%2Freleases~875395","triplet_id":"openstack%2Freleases~master~I2ee34c05776f83a2223858d2c342e5042c384a15","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2023-02-27 13:27:34.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-03 11:24:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2ee34c05776f83a2223858d2c342e5042c384a15","subject":"Release ceilometer RC1 for Antelope","status":"MERGED","created":"2023-02-27 11:04:29.000000000","updated":"2023-03-03 11:24:03.000000000","submitted":"2023-03-03 11:24:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875395-antelope-rc1-deadline","meta_rev_id":"9dbc9707a49dea05a55ac53a3c468e2c41c6297e","_number":875395,"virtual_id_number":875395,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-03 11:24:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-03-03 11:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2023-03-03 10:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-03-03 11:03:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 11:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 11:18:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 12:24:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-03-03 11:03:13.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"18fc5aeca15cc73fb50c962126a312a8a616980e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 11:04:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbdef191715f95682ae94474b67e64809530c179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 11:18:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d243cb35057c49268e6b79826dc572c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c93fed50fd4468a67c6fcc53260828 : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed14b8deaf534894b177a1ed6777d822 : SUCCESS in 5m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0c53bdb5e3354973afc7bb0ccbd8cfea : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b026a0883ff6d6a5b3886e597183e15807c0195c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-02-27 13:27:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"19be8cf639c0deee9d480c96fda27d4ef50d6e84","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 13:28:24.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"365cf013193beaa5d6ff53aa48233afc7f82e4b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 13:59:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0a903d707a16320a027a0c231a1eeb1ef75c99a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:23:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e08b542aede2425ebc4504881db0201a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b54ccb2639c94980a13cddc636e98caa : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce747778f1ea4169b34ed758cf21c48e : SUCCESS in 5m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4310b90c99704ddebf72c46d4f5d9aa9 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"2582c304d082f3edaaf3c0a2d4342a2ef365ff8d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-03 10:42:03.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"562918238dd52acc0eeb9c5b8e8e05d783d3e773","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-03 11:03:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"92b3a516a34e32442968e35314f385d1c0d92819","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 11:03:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d660335e6ccf2c189899cbc9e120879587f7e873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 11:24:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f127cac33d654e289cc86750b05da3df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fdaac6d90304596bcf6bb0be31a13d8 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac8905e68d41404b9bc8f30fd0a0e421 : SUCCESS in 9m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/181ec41e973f4092bba31d4e45f06feb : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9dbc9707a49dea05a55ac53a3c468e2c41c6297e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 11:24:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98fda6efe4e67440aa087f0390c4b3b4fb28b849","revisions":{"e2783570c6222b2dda0cd91ed6cc5b390190e01a":{"kind":"REWORK","_number":1,"created":"2023-02-27 11:04:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/95/875395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/875395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/875395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/875395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/875395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/875395/1"}}},"commit":{"parents":[{"commit":"bda9d1613b39b8521d5e2f66b780080ae47d845a","subject":"Merge \"Release networking-baremetal 6.1.0 for antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bda9d1613b39b8521d5e2f66b780080ae47d845a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:04:25.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:04:25.000000000","tz":60},"subject":"Release ceilometer RC1 for Antelope","message":"Release ceilometer RC1 for Antelope\n\nThis proposed a RC1 release for ceilometer to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 19.0.0..ccce4155\n0118742b Remove SNMP metrics from default polling\na58e1b1c Imported Translations from Zanata\n39d0ef6c Imported Translations from Zanata\ncb7084ed Fix gnocchi install from git\ne83a6802 Imported Translations from Zanata\ned55b5f4 Add TCP publisher\n8434d14e Make tox.ini tox 4.0 compatible\n8810c4bc Imported Translations from Zanata\n463594b2 NoUniqueMatch: ClientException on Gnocchi publisher\n264229ea Fix barbican credentials retrieval\n2a267388 Tests the sample name\n7a507457 Fix OutgoingBytesDeltaPollster sample name\nb4a2801e Change oslo_cache implementation\n060a1345 Improve logging for Gnocchi publisher\n9fe3674e Imported Translations from Zanata\n704fa74e Post individual JSONs instead of the whole list\nf0730df4 Update the installation guide for RHEL8/CentOS8\n8fcb3804 Imported Translations from Zanata\n44caf759 Fix the handle of plain text outputs\n7e545b0f Remove lingering queue declaration\nb7c2c7ca Properly handle \u0027resource_id\u0027 as None for Gnocchi publisher\n090a0eb6 Add support to namespaces on dynamic pollsters\n1d5aab1c Imported Translations from Zanata\nbff9879e Add extra metadata fields skip\ncb448a1d Add support to host command dynamic pollster definitions\n79454d6b Add user/project names to polled samples\n31ca74fa Switch to 2023.1 Python3 unit tests and generic template name\nc4dca498 Update master for stable/zed\n8e57a3a6 Exclude metrics from \"service\" project\n26d569de Remove unnecessary logic for missing oslo.cache\n225f1cd7 Add response handlers to support different response types\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I2ee34c05776f83a2223858d2c342e5042c384a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e2783570c6222b2dda0cd91ed6cc5b390190e01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e2783570c6222b2dda0cd91ed6cc5b390190e01a"}]},"branch":"refs/heads/master"},"98fda6efe4e67440aa087f0390c4b3b4fb28b849":{"kind":"REWORK","_number":2,"created":"2023-02-27 13:59:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/95/875395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/875395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/875395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/875395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/875395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/875395/2"}}},"commit":{"parents":[{"commit":"bda9d1613b39b8521d5e2f66b780080ae47d845a","subject":"Merge \"Release networking-baremetal 6.1.0 for antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bda9d1613b39b8521d5e2f66b780080ae47d845a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:04:25.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 13:59:04.000000000","tz":0},"subject":"Release ceilometer RC1 for Antelope","message":"Release ceilometer RC1 for Antelope\n\nThis proposed a RC1 release for ceilometer to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 19.0.0..ccce4155\n0118742b Remove SNMP metrics from default polling\na58e1b1c Imported Translations from Zanata\n39d0ef6c Imported Translations from Zanata\ncb7084ed Fix gnocchi install from git\ne83a6802 Imported Translations from Zanata\ned55b5f4 Add TCP publisher\n8434d14e Make tox.ini tox 4.0 compatible\n8810c4bc Imported Translations from Zanata\n463594b2 NoUniqueMatch: ClientException on Gnocchi publisher\n264229ea Fix barbican credentials retrieval\n2a267388 Tests the sample name\n7a507457 Fix OutgoingBytesDeltaPollster sample name\nb4a2801e Change oslo_cache implementation\n060a1345 Improve logging for Gnocchi publisher\n9fe3674e Imported Translations from Zanata\n704fa74e Post individual JSONs instead of the whole list\nf0730df4 Update the installation guide for RHEL8/CentOS8\n8fcb3804 Imported Translations from Zanata\n44caf759 Fix the handle of plain text outputs\n7e545b0f Remove lingering queue declaration\nb7c2c7ca Properly handle \u0027resource_id\u0027 as None for Gnocchi publisher\n090a0eb6 Add support to namespaces on dynamic pollsters\n1d5aab1c Imported Translations from Zanata\nbff9879e Add extra metadata fields skip\ncb448a1d Add support to host command dynamic pollster definitions\n79454d6b Add user/project names to polled samples\n31ca74fa Switch to 2023.1 Python3 unit tests and generic template name\nc4dca498 Update master for stable/zed\n8e57a3a6 Exclude metrics from \"service\" project\n26d569de Remove unnecessary logic for missing oslo.cache\n225f1cd7 Add response handlers to support different response types\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I2ee34c05776f83a2223858d2c342e5042c384a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98fda6efe4e67440aa087f0390c4b3b4fb28b849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98fda6efe4e67440aa087f0390c4b3b4fb28b849"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
