)]}'
{"id":"openstack%2Freleases~875403","triplet_id":"openstack%2Freleases~master~I5c1832dd099efdac84e533233ee9230f55fb964e","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2023-02-28 01:01:50.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-28 13:21:56.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2023-02-28 13:21:56.000000000","reason":"Change was submitted"},"13629":{"account":{"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"},"last_update":"2023-02-28 13:21:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c1832dd099efdac84e533233ee9230f55fb964e","subject":"Release cyborg RC1 for Antelope","status":"MERGED","created":"2023-02-27 11:20:59.000000000","updated":"2023-02-28 13:21:56.000000000","submitted":"2023-02-28 13:21:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875403-antelope-rc1-deadline","meta_rev_id":"8b0b4a4845749d0cb7c0705a7fd92abe1a2a6bc8","_number":875403,"virtual_id_number":875403,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 13:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2023-02-28 01:01:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2023-02-28 12:48:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-02-28 10:36:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2023-02-28 12:48:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-02-28 01:02:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 11:21:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 11:32:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 12:24:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"},"state":"REVIEWER"},{"updated":"2023-02-27 12:24:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2023-02-27 12:24:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2023-02-28 12:48:38.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"c9c2eedc59588518fd2b0253886bb95cdfceab99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 11:20:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb23b25daa6dcd8729c9a7031f36c15ac77d80b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 11:32:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9350f72837b1440c8e5ca8c78a082db9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/997c258dae834ad99c809e513d63cda5 : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab5ee17069174c5f85c4b0a674a9dd0a : SUCCESS in 5m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/38068185cc1643d387e87f41046cbb3b : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"010c3c90d4641b616fad2d1a7a36fb743e0b5f5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:32:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"938df9eeb26939217cdd4942c6df224b060b1d2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:51:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8385926fdf74f638a398849ec1b5798\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ef6e1489fd4ca6b2d93eba5cc64d83 : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/122c2d9099b940e2a191704f4b9ebbfd : SUCCESS in 5m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fe56ea1955bb47559c1ba033e89ea419 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"cde821c057e98329cd8bd07f44409022d91dbb98","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-02-28 01:01:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7b03abb1ed4e17eadd38f45bc1ac24944f763c44","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 01:02:34.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"21aff709c8bead4b4ca1d5d2860942e9b7142fee","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-28 10:36:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"88d5669b7bddf281594159f64ec66ccc0d173524","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-02-28 12:48:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"63b7e24878ca4704cf5a601140d37c90a35953fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 12:49:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"437682620658c4f3807e79bd0af76c4f41ef90f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 13:21:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ee3427fc3314220ad117d3c68988084\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c046e3079a1d43229b8a34254e31cc4c : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/addd0ec2e6a141bb9f299eb6d81e285d : SUCCESS in 8m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3468cb819e34d8d996ecec3b31a48a5 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":2},{"id":"8b0b4a4845749d0cb7c0705a7fd92abe1a2a6bc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 13:21:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"74fbfd34011520634017e88b700f6d939b42eeb0","revisions":{"be99ce9d0de3b91e35ca0e993341d8e2b79dc959":{"kind":"REWORK","_number":1,"created":"2023-02-27 11:20:59.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/03/875403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/03/875403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/875403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/875403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/875403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/03/875403/1"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:20:55.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:20:55.000000000","tz":60},"subject":"Release cyborg RC1 for Antelope","message":"Release cyborg RC1 for Antelope\n\nThis proposed a RC1 release for cyborg to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 9.0.0..eafcc2f\neafcc2f Fix the wrong use of variable assertions in unit tests\n0550592 When asserting the number of members obtained in unit test deployables, should use assertEqual\n99858d0 fix function usage errors\n3f052b6 fix function usage errors\n1a0a92a add attributes delete api ut\nc5c984b add attributes get by deployable id and key api ut\n954ee30 fix the spell error\n510486d add attributes create api ut\nce622a6 add attributes  get one by key api ut\nf2e24e3 add attributes  get all api ut\ne7a2246 add attributes  get one by deployable_id api ut\n58e46e9 Support create attribute api\n08f63a8 support get attribute list filter by deployable_id and key\n8934856 add attributes  get one by uuid api ut\n3ba22d6 Fix attribute get api\n61ade66 Fix docs: change device_profile_uuid to body in get api\n8fbd52d change text type to uuid for instance, device_rp_uuid, project_id and arq\n43c8367 Add placement put rp traits api exception ut\nb715d24 Add placement put rp traits api ut\n1df18be add attributes api ut file\nca21fe1 Add placement ensure traits api exception ut\n4275e49 Add placement ensure traits api ut\ncf70b79 Add placement get rp traits api exception ut\n8554e50 Add placement get rp traits api ut\n0ceab2e Add placement delete api exception ut\n9edb0e0 Add placement delete api ut\n533e343 Add placement post api exception ut\ncb1331a Improve the api-ref docs for v1 version\nc9db95a Add placement put api ut\nf2a79aa Add placement post api exception ut\n8677639 Add placement post api ut\n9012520 cyborg Fix tox4 error\n5c504ec Add placement get exception ut\nac4fb54 Add ut for placement API\na0788f7 Get an attribute by it\u0027s deployable_id.\nfaa67c3 Add delete method for attributes\n2eb0be1 Add get_one for attributes.\n962ceed Add attribute_list_api\n373afa3 Add project id sample in the request body\n2bfa663 Use py3 as the default runtime for tox\nb973622 Switch to 2023.1 Python3 unit tests and generic template name\n548c5f8 Update master for stable/zed\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I5c1832dd099efdac84e533233ee9230f55fb964e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be99ce9d0de3b91e35ca0e993341d8e2b79dc959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be99ce9d0de3b91e35ca0e993341d8e2b79dc959"}]},"branch":"refs/heads/master"},"74fbfd34011520634017e88b700f6d939b42eeb0":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:32:26.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/03/875403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/03/875403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/875403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/875403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/875403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/03/875403/2"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:20:55.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 14:32:25.000000000","tz":0},"subject":"Release cyborg RC1 for Antelope","message":"Release cyborg RC1 for Antelope\n\nThis proposed a RC1 release for cyborg to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 9.0.0..eafcc2f\neafcc2f Fix the wrong use of variable assertions in unit tests\n0550592 When asserting the number of members obtained in unit test deployables, should use assertEqual\n99858d0 fix function usage errors\n3f052b6 fix function usage errors\n1a0a92a add attributes delete api ut\nc5c984b add attributes get by deployable id and key api ut\n954ee30 fix the spell error\n510486d add attributes create api ut\nce622a6 add attributes  get one by key api ut\nf2e24e3 add attributes  get all api ut\ne7a2246 add attributes  get one by deployable_id api ut\n58e46e9 Support create attribute api\n08f63a8 support get attribute list filter by deployable_id and key\n8934856 add attributes  get one by uuid api ut\n3ba22d6 Fix attribute get api\n61ade66 Fix docs: change device_profile_uuid to body in get api\n8fbd52d change text type to uuid for instance, device_rp_uuid, project_id and arq\n43c8367 Add placement put rp traits api exception ut\nb715d24 Add placement put rp traits api ut\n1df18be add attributes api ut file\nca21fe1 Add placement ensure traits api exception ut\n4275e49 Add placement ensure traits api ut\ncf70b79 Add placement get rp traits api exception ut\n8554e50 Add placement get rp traits api ut\n0ceab2e Add placement delete api exception ut\n9edb0e0 Add placement delete api ut\n533e343 Add placement post api exception ut\ncb1331a Improve the api-ref docs for v1 version\nc9db95a Add placement put api ut\nf2a79aa Add placement post api exception ut\n8677639 Add placement post api ut\n9012520 cyborg Fix tox4 error\n5c504ec Add placement get exception ut\nac4fb54 Add ut for placement API\na0788f7 Get an attribute by it\u0027s deployable_id.\nfaa67c3 Add delete method for attributes\n2eb0be1 Add get_one for attributes.\n962ceed Add attribute_list_api\n373afa3 Add project id sample in the request body\n2bfa663 Use py3 as the default runtime for tox\nb973622 Switch to 2023.1 Python3 unit tests and generic template name\n548c5f8 Update master for stable/zed\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I5c1832dd099efdac84e533233ee9230f55fb964e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/74fbfd34011520634017e88b700f6d939b42eeb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/74fbfd34011520634017e88b700f6d939b42eeb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
