)]}'
{"id":"openstack%2Freleases~875413","triplet_id":"openstack%2Freleases~master~I023233da893c04b4891ea38c3aec13ec50155574","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2023-03-07 12:27:10.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-07 12:27:10.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-07 12:27:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I023233da893c04b4891ea38c3aec13ec50155574","subject":"Release heat RC1 for Antelope","status":"MERGED","created":"2023-02-27 11:35:08.000000000","updated":"2023-03-07 12:27:10.000000000","submitted":"2023-03-07 12:27:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875413-antelope-rc1-deadline","meta_rev_id":"0ac5088e7e4121510f61fcc02c112a83050e5d4d","_number":875413,"virtual_id_number":875413,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 12:27:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-03-07 11:39:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-03-07 12:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-03-07 12:13:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-03-07 11:07:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 11:35:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 11:59:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 12:22:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2023-02-27 12:22:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-03-06 08:45:16.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2023-03-07 11:39:09.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-03-07 12:13:38.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c789cdcfef9649010459caf3e3f79a7ff016e1d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 11:35:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b08cc7010cb594d03958e109c02ada6da5e494e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 11:59:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee4d6d43770d466c8a4979469d3d9cca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d21d8ca3894941a5d6ebff7812487a : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa9288535255489189e1bedf2acdefb4 : SUCCESS in 5m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f35a0c3a6ab6448691e59cb9008fdd52 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8a808275d11339fda5108a2e6fdefc65f742fd9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:39:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"77754aadb948db166dc4307afe67e0d3ac01726e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:01:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30cd0416855e4c5a953ccbe291993de1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7458991a296406780888b3453c59ad1 : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b3a6f6bad8042bca1e5bfec48520b07 : SUCCESS in 5m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5b2ee25561644edd9975ed350e8f9101 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0d6a86b3a2aeea12ff94c6e7968acdf1bdce7d85","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-02 06:49:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"136cae5a216df24e12ceed14f643713cb1cabaf5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-06 08:45:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48def35eacef72d0130d2b81f715be1b99b02671","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-07 11:06:36.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7cc4ddbf5fc66575bd767d53eb3f0613f66d43d9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 11:07:34.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"9ecf982a575d2548c238bf3eaa9739382acce8c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 11:25:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a507ea6fad4a464cbdf386d2b14e932d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7177f2071870434eb34dbe64926ca749 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/347e5338353842c58f208a8af24e81a2 : SUCCESS in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c644a7d1e8a54f3cb746c2fb2cd253d3 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"bb9e8907a633e18e00e123e6ee3d890469e3343f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-07 11:39:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c2c7bc0727b0b52b95c30173a9469dd13df93170","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-07 12:13:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82a78710a7959687ab5a02f829bba1efa38a0c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 12:13:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"11453ce53c74c93234398c965cbf7299b83a240d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 12:27:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5848cfb51a924c2da83f057deb22975d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc3e8a7c8b24f1781f32c869622c3cc : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df8967e8144452a9ebb391a53cb2cdb : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a613e888f29a404a9f17cd482f419899 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":3},{"id":"0ac5088e7e4121510f61fcc02c112a83050e5d4d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 12:27:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2994b4eda975d77887d8ba85ee86546022d57bbc","revisions":{"5143122aa4504cb9b72de9084439bfca95524d6f":{"kind":"REWORK","_number":1,"created":"2023-02-27 11:35:08.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/13/875413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/13/875413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/13/875413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/13/875413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/13/875413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/13/875413/1"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:35:04.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:35:04.000000000","tz":60},"subject":"Release heat RC1 for Antelope","message":"Release heat RC1 for Antelope\n\nThis proposed a RC1 release for heat to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 19.0.0.0rc1..604111b0b\n604111b0b Fix setting the tempest virtual env constraints env var\n8e2286c9b Fix unit tests for python-barbicanclient 5.5.0\n85480f94d Fix an assert statement\n29eaf4ef7 Use new get_rpc_client API from oslo.messaging\na0ff4b98b Decode UTF-8 body data in SwiftSignal\n1305a3152 Honor \u0027hidden\u0027 parameter in \u0027stack environment show\u0027 command\nb452ea708 Skip test_stack_update_with_replacing_userdata\n53fa8c374 Explicitly pass error kwarg in nested StackValidationFailed\n500201cb7 Skip test_cancel_update_server_with_port\n9407b4897 Ignore resource_data decryption errors\nceef7869e Don\u0027t load all resources for policy check\n5451b9854 Imported Translations from Zanata\n0aa0b3825 Update stestr args for inclusive language\n56d99bf65 Improve map_merge logging\nddbf9db52 Set cache_ok to avoid further SAWarning\n0ea1ee087 Retry to detach volume if nova returned error 409\n13fbd367e Fix get_attr syntax in template guide\n87bdbfdca Drop implementation for OS::Magnum::Bay/BayModel\nba29a7089 Switch to 2023.1 Python3 unit tests and generic template name\n1b42c7296 Update master for stable/zed\naf5ffbe01 Replace deprecated Query.as_scalar()\ne17f68a83 Deploy healthcheck middleware as app instead of filter\n520e2389d Allow deleting user_creds when can\u0027t be decrypted\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I023233da893c04b4891ea38c3aec13ec50155574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5143122aa4504cb9b72de9084439bfca95524d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5143122aa4504cb9b72de9084439bfca95524d6f"}]},"branch":"refs/heads/master"},"0dbe6bec943accd2db676aecc16a17792b70cca1":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:39:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/13/875413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/13/875413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/13/875413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/13/875413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/13/875413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/13/875413/2"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:35:04.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 14:39:49.000000000","tz":0},"subject":"Release heat RC1 for Antelope","message":"Release heat RC1 for Antelope\n\nThis proposed a RC1 release for heat to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 19.0.0.0rc1..604111b0b\n604111b0b Fix setting the tempest virtual env constraints env var\n8e2286c9b Fix unit tests for python-barbicanclient 5.5.0\n85480f94d Fix an assert statement\n29eaf4ef7 Use new get_rpc_client API from oslo.messaging\na0ff4b98b Decode UTF-8 body data in SwiftSignal\n1305a3152 Honor \u0027hidden\u0027 parameter in \u0027stack environment show\u0027 command\nb452ea708 Skip test_stack_update_with_replacing_userdata\n53fa8c374 Explicitly pass error kwarg in nested StackValidationFailed\n500201cb7 Skip test_cancel_update_server_with_port\n9407b4897 Ignore resource_data decryption errors\nceef7869e Don\u0027t load all resources for policy check\n5451b9854 Imported Translations from Zanata\n0aa0b3825 Update stestr args for inclusive language\n56d99bf65 Improve map_merge logging\nddbf9db52 Set cache_ok to avoid further SAWarning\n0ea1ee087 Retry to detach volume if nova returned error 409\n13fbd367e Fix get_attr syntax in template guide\n87bdbfdca Drop implementation for OS::Magnum::Bay/BayModel\nba29a7089 Switch to 2023.1 Python3 unit tests and generic template name\n1b42c7296 Update master for stable/zed\naf5ffbe01 Replace deprecated Query.as_scalar()\ne17f68a83 Deploy healthcheck middleware as app instead of filter\n520e2389d Allow deleting user_creds when can\u0027t be decrypted\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I023233da893c04b4891ea38c3aec13ec50155574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0dbe6bec943accd2db676aecc16a17792b70cca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0dbe6bec943accd2db676aecc16a17792b70cca1"}]},"branch":"refs/heads/master"},"2994b4eda975d77887d8ba85ee86546022d57bbc":{"kind":"REWORK","_number":3,"created":"2023-03-07 11:06:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/875413/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/13/875413/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/13/875413/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/13/875413/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/13/875413/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/13/875413/3"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:35:04.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-07 11:06:36.000000000","tz":0},"subject":"Release heat RC1 for Antelope","message":"Release heat RC1 for Antelope\n\nThis proposed a RC1 release for heat to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 19.0.0.0rc1..604111b0b\n604111b0b Fix setting the tempest virtual env constraints env var\n8e2286c9b Fix unit tests for python-barbicanclient 5.5.0\n85480f94d Fix an assert statement\n29eaf4ef7 Use new get_rpc_client API from oslo.messaging\na0ff4b98b Decode UTF-8 body data in SwiftSignal\n1305a3152 Honor \u0027hidden\u0027 parameter in \u0027stack environment show\u0027 command\nb452ea708 Skip test_stack_update_with_replacing_userdata\n53fa8c374 Explicitly pass error kwarg in nested StackValidationFailed\n500201cb7 Skip test_cancel_update_server_with_port\n9407b4897 Ignore resource_data decryption errors\nceef7869e Don\u0027t load all resources for policy check\n5451b9854 Imported Translations from Zanata\n0aa0b3825 Update stestr args for inclusive language\n56d99bf65 Improve map_merge logging\nddbf9db52 Set cache_ok to avoid further SAWarning\n0ea1ee087 Retry to detach volume if nova returned error 409\n13fbd367e Fix get_attr syntax in template guide\n87bdbfdca Drop implementation for OS::Magnum::Bay/BayModel\nba29a7089 Switch to 2023.1 Python3 unit tests and generic template name\n1b42c7296 Update master for stable/zed\naf5ffbe01 Replace deprecated Query.as_scalar()\ne17f68a83 Deploy healthcheck middleware as app instead of filter\n520e2389d Allow deleting user_creds when can\u0027t be decrypted\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I023233da893c04b4891ea38c3aec13ec50155574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2994b4eda975d77887d8ba85ee86546022d57bbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2994b4eda975d77887d8ba85ee86546022d57bbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
