)]}'
{"id":"openstack%2Freleases~875439","triplet_id":"openstack%2Freleases~master~I850a8265be9b43753a97210513b02648c9e01424","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-28 12:09:04.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2023-03-01 11:12:23.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-01 11:12:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I850a8265be9b43753a97210513b02648c9e01424","subject":"Release networking-odl RC1 for Antelope","status":"MERGED","created":"2023-02-27 11:55:52.000000000","updated":"2023-03-01 11:12:23.000000000","submitted":"2023-03-01 11:12:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875439-antelope-rc1-deadline","meta_rev_id":"e4a452a33d8ba8dd495a5cbc000cdd18120bc3b8","_number":875439,"virtual_id_number":875439,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-01 11:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-02-28 12:09:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-02-28 12:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-03-01 11:00:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-28 13:23:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-03-01 11:00:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-02-28 12:09:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 11:56:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 12:22:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2023-02-27 12:22:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-02-27 12:26:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 12:52:54.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-02-28 13:23:00.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-03-01 11:00:36.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"41e99f52db60a22e66eff82309ccfa38420ed355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 11:55:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1176dd9243ec3f82fb656a1064c744891ee8ff61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:26:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1078732ff09a45d5be0f2f2ed4527331\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd98530304f4e4fabaa652e5788bc4d : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04a773d73a7f4bbb95bc55e677742b98 : SUCCESS in 6m 10s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0ee8c6afa77c4faab0df73d11afe3d9c : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7b2cc8d375027b8b1b516139ebd1df9b977e9dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:21:32.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b0f9457b499a02b605ce62c8a469539e067502e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:48:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd31059021354084b3c5105dc3c5738e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c768b4aeb1d44b13b12b1c51548d6ae2 : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af95638ac2da4841bbe536d0252fbd51 : SUCCESS in 5m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ce88667795164a2bb32248d7b9ed1375 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"22960a980435bd2e74438f4963b41dfea827cbcd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 12:09:04.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1587ed959c3b07c8698930edd860528fd79b6106","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 12:09:59.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"a0921493e2ad083e9edd1bdb7a1b41432958495e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-02-28 12:52:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f20eabf73d268bcccb2444a61631240ae77f1511","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-28 13:23:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"73988778e9efbfa391775c2aac6d3d9b8e4a6a57","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-01 11:00:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74c44b86d852d39f1dfdc89f8c4bb3e32d7f4af2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 11:00:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"264c36002d665650956fc02d4edc4aa311735a87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 11:12:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/089579bbf95842219a3385bb850e0413\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce7658a368547058976836a407fd6e6 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d47d9b65ab4b18a5797625eb8cd9eb : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbded5325ff64fbb8964719fb85411ad : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"e4a452a33d8ba8dd495a5cbc000cdd18120bc3b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 11:12:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f35ea8f08baa3e1d5b37e7f1e68c758257e3a496","revisions":{"1379460fcb3a6eb2a4a43ea1ccab379c5d642a69":{"kind":"REWORK","_number":1,"created":"2023-02-27 11:55:52.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/39/875439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/39/875439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/875439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/875439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/875439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/39/875439/1"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:55:48.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:55:48.000000000","tz":60},"subject":"Release networking-odl RC1 for Antelope","message":"Release networking-odl RC1 for Antelope\n\nThis proposed a RC1 release for networking-odl to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 21.0.0..d8316e6f5\nd8316e6f5 Tox4: add allowlist_externals where necessary\n2e2588c4c Remove lib/neutron-legacy leftovers\nb92b98643 Imported Translations from Zanata\neecb7125a CI: Increase the timeout of functional jobs\ne44ff5114 Imported Translations from Zanata\n0f0e4f936 Switch to 2023.1 Python3 unit tests and generic template name\n1d8e3fd8f Update master for stable/zed\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I850a8265be9b43753a97210513b02648c9e01424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1379460fcb3a6eb2a4a43ea1ccab379c5d642a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1379460fcb3a6eb2a4a43ea1ccab379c5d642a69"}]},"branch":"refs/heads/master"},"f35ea8f08baa3e1d5b37e7f1e68c758257e3a496":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:21:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/39/875439/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/39/875439/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/875439/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/875439/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/875439/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/39/875439/2"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 11:55:48.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 14:21:32.000000000","tz":0},"subject":"Release networking-odl RC1 for Antelope","message":"Release networking-odl RC1 for Antelope\n\nThis proposed a RC1 release for networking-odl to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 21.0.0..d8316e6f5\nd8316e6f5 Tox4: add allowlist_externals where necessary\n2e2588c4c Remove lib/neutron-legacy leftovers\nb92b98643 Imported Translations from Zanata\neecb7125a CI: Increase the timeout of functional jobs\ne44ff5114 Imported Translations from Zanata\n0f0e4f936 Switch to 2023.1 Python3 unit tests and generic template name\n1d8e3fd8f Update master for stable/zed\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I850a8265be9b43753a97210513b02648c9e01424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f35ea8f08baa3e1d5b37e7f1e68c758257e3a496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f35ea8f08baa3e1d5b37e7f1e68c758257e3a496"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
