)]}'
{"id":"openstack%2Freleases~875446","triplet_id":"openstack%2Freleases~master~Iefe51ff7ca7e451d4149ad8886896e990d7066e2","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-02 11:20:54.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-02 10:14:33.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2023-03-02 11:20:54.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-02 11:10:35.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"Iefe51ff7ca7e451d4149ad8886896e990d7066e2","subject":"Release neutron RC1 for Antelope","status":"MERGED","created":"2023-02-27 12:03:04.000000000","updated":"2023-03-02 11:20:54.000000000","submitted":"2023-03-02 11:20:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875446-antelope-rc1-deadline","meta_rev_id":"f0f26df7dfbb9d420470234eb64d7db0cd66be44","_number":875446,"virtual_id_number":875446,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 11:20:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-03-02 08:42:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-02 09:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-03-02 10:14:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-03-02 09:49:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-03-02 11:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-03-02 11:10:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-03-02 08:42:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 12:03:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 12:22:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2023-02-27 12:22:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-02-27 12:29:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-01 13:21:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-02 09:49:15.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-03-02 09:49:20.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"97791608ecb8f142334ac012a49d5dbbdd68bf76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 12:03:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fecfd732f060c0c9f511f99b375cf300984c6703","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:29:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba203313df9349bcac760971f196837a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a9b2bb17afa4988bdab2b1189fb704b : SUCCESS in 7m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec31bd142301489bbbcc2f3cdd7ff4dd : SUCCESS in 4m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/85ca1ab3820f43ffa7ce12568e3600c8 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6b14e9f950c202862f4447f3695d64043fe2df83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:18:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c293f410c86f5a89e66cbc724b362f2bc9072d11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:29:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/324fb89e3aea438c80ac3fb5ece857cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ddf1ff65e4d418aad621247c2edad95 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e13bfdf2b52f49959514ff1e3d04fa48 : SUCCESS in 5m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c169f861730a4dd2a53fa8a8fc79cefe : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b837cb0cbf15faf20db491d99817faa6b8247ca9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 11:20:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39433f380d5dfe3a3cfbc03508c55e3b251c3b11","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-01 14:43:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6e09fe512aff3691aa2695b896a6034cc47eabf3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 15:25:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1efcb4fc4931c39bb5668547c36ba4a0b7aa16ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 08:42:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"aea6330d6f99c9feed5f548973462bf4b37fc1b8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 08:42:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b4ec8155e29bcb96a4b81f533381636d7d9c92ab","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 08:42:57.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"b68019baa28a4341198fa3d4b597c8cac7088698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 09:05:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02a57d1066a049cdb8bcc329dac8f53e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3669ef1e25614078bbba599db5e74201 : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0f8452e832a49cb85f3637f38fbba48 : SUCCESS in 5m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fa1c116da24f41bab937ac4d0324711b : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"7ad6a4a63156deb273c6ad5ead230e3713fcd6d3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-02 09:49:15.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c01acb0720e39390f5191377d68fbb808d47004","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-02 09:49:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6a4a94d79a9787a757013b9fb775d5dc54fb0703","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-02 10:14:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dbb311fd143623e7f206c955792c63d38c897f6f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-02 11:10:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"78a595a5d80a20a69edd4574ac1091c8dd9cc0b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 11:11:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"95377fb76809dbf30772cb79b6a4e044e4cc4b16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 11:20:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ccb98aa872940349145940df5a4bf4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bc694df9af74b35aff0bb67cbce28c5 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3377eb88ea4ecca9db277d199f042f : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f9dcfab3c82474ca1cd4fa6a2ed7421 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":3},{"id":"f0f26df7dfbb9d420470234eb64d7db0cd66be44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 11:20:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"66b571ee5283e406ccb31b7c48235fbdda20b041","revisions":{"6b783e08c84dfe7ecf92a1d97778659e598fd270":{"kind":"REWORK","_number":1,"created":"2023-02-27 12:03:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/875446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/875446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/875446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/875446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/875446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/875446/1"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:03:00.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:03:00.000000000","tz":60},"subject":"Release neutron RC1 for Antelope","message":"Release neutron RC1 for Antelope\n\nThis proposed a RC1 release for neutron to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 21.0.0.0rc1..b30d7a2d03\naa40aef70f Mark fullstack TestMultiSegs.test_multi_segs_network as unstable\nf991fc2fbe [ovn][ipv6] Skip test_update_delete_extra_route\n51d449babf Move tox-py311 job override into job-templates.yaml\nc61ce3447e Format correctly (dialect\u003dmac_unix_expanded) the MAC addresses\n502c0253de [ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class\ne9ce025a69 [dhcp agent] Fix clean_devices\nd052da85cf Mark ``test_port_creation_and_deletion`` as unstable\n5e15c4a4d4 Allow other kwargs in ``DhcpBase.disable``\n5dac5f8c6c [ovn-ipv6] Temporary exclude test_hotplug_nic\n67a232fe71 Remove tripleo related jobs from neutron queues\n595c387efb [UT] Mock the retry decorator when creating duplicated NDPProxy\nf9880c2013 Increase tox-py311 timeout to 3600 seconds\n24152f2019 Add debugging exception messages in ``test_ovs_and_ovs_events``\nfce516e346 [OVN] Bump the port revision number in trunk driver\nd83e243c6d [OVN] Add missing LSP device_owner info in trunk driver\n76f3b75ad6 [OVN] Skip ports with VNIC type direct and no port binding\n6df75ca696 Deprecate allow_stateless_action_supported config option\n5250598c80 Improve scheduling L3/DHCP agents, missing lower binding indexes\n8f7cc690a9 Remove ``L3Scheduler._bind_routers`` method\n47fef55e25 Add a method to retrieve and register \"rpc_workers\" config knob\n359030f741 Do not log full network on reload_allocations\n4e71675ddb rbacs: add functionl test that asserts behavior on networks/users\n5c697b8d60 allow manila ports to do multiple port binding for ML2\nc15313ea21 Removed duplicated job call in experimental queue\nc93c31852b Adapt deploy_rootwrap filters path for tox4\n61b2917a3e Filter subnets by \"enable_dhcp\" flag using the correct type\n8be42f6d72 Enable qos extension_driver in ovn_db_sync\n3c2af924d9 Add CI jobs using SQLAlchemy master branch\nabfa8940b1 [OVN] Fix the OVN Agent config file location\n54eff20a72 [OVN] New OVN Neutron Agent extension: QoS for HWOL\n07fd743dc9 Remove extra call to config_logging() for neutron-server\ne98a268de4 Propose replacement of ORM from_self()\n1c27d8b5b4 [Secure RBAC] Fix policy to get flavors\n17be2505d1 [Secure RBAC] Allow network owner to delete ports plugged to the network\n5d505e7b7c [Secure RBAC] Add shared_security_groups rule\n221cf9497f [Secure RBAC] Add shared_qos_policy rule\nf7e31b4c05 [OVN] Allow logging all traffic related to an ACL\n267ad86291 Add tox Python3.11 job to the testing queues\n83da1e6d79 Improve message for subnet gateway out of host IP addresses range\n7cb481a3dc Allow disable stateful security group extension on older OVN\n48c3621292 Imported Translations from Zanata\n3727103ce7 Fix get_link_devices() with index argument\n67e616b238 Never raise an exception in notify()\nd0c7bb653a [OVN] Implementation of OVN Neutron Agent\nc609084b59 Add doc note on nf_conntrack module requirement\n0ae432e717 Fullstack: Wait placement process fixtrue to really stop\n910574a5cc Use SLQAlchemy ORM \"relationship\" instead of \"relation\"\n1d22cfa99a Change flag check order in wait_until_address_ready()\ned68ba4a4c Do not ignore attributes in bulk port create\nd73f75c551 [API] Add API extension and definition for default SG rules\n1a3bdff18a Honor debug mode in keepalived-state-change script logs\nec438577bd Use dhcpcd client in the tempest slow jobs\nf235ce8791 Add new debugging logs for ``ARPSpoofTestCase``\n78c3d52bb4 [OVN] Introduce the new OVN Neutron Agent\nfeb9f2b21c ovn-migration: Stop neutron server while running db sync\n2495ad2856 Avoid failure of get_device_details if net from cache is empty\nb6750fb2b8 [OVN] Add vnic_type and binding profile capabilities to LSP info\na3bf74baa5 Ignore reno artefacts (RELEASENOTES.rst and reno.cache)\n1d84aaf877 Bump the minimum pyroute2 version to 0.7.3\n463c3df4cf [ovn] Do not create empty default route when empty gateway_ip\n70a86637e7 Allow shared resources between physical and tunnelled networks\nf28e416599 Add 3 secs to wait for keepalived state change\nd7f68a0ce7 Set proxy_delay to one when using proxy ARP\nd0e64c6617 Add auto-allocated-topology delete example to docs\n3354b43d5a Increase fullstack job\u0027s timeout\ne2097157bb Allow multiple IPv6 ports on router from same network ml2/ovs+dvr\nde0918cea3 Add debug logs to OVSInterfaceDriver.plug_new\n7751ac32a7 Add info on segments extension to contrib guide\ned048638f4 Add text to WSGI config guide for debugging\nc362016620 Ensure that MAC address of the device is set correctly\n3e1e2d63b3 Enable rpc notifications only when rpc_workers \u003e\u003d 1\n1c89db495c [OVN][FT] Also update nb_cfg along with nb_cfg_timestamp\nd59bf5ff6c Temporary update envlist for tox4\n25ec6e7e4f Set ip_nonlocal_bind to 1 for HA routers and DVR snat\n07fd0e101b Revert \"Documentation for adding extra port conflict fix\"\n6501ebc6ca Documentation for adding extra port conflict fix\nf759915ab0 Fix default value for MTUs, when not provided\n02030f037a ovs-dpdk support in neutron-sanity-check\n9f9478071e Bump Cirros version used in CI jobs to 0.6.1\nbee07defac Track all interfaces in Keepalived\n26da863ca6 Remove deprecated opt keepalived_use_no_track\n4fa85b9a99 Fix incorrect mysql examples\nedf48e46a1 Improve agent provision performance for large networks\n5e19eaadbb [OVN] Fix availability zones changes check\n3ebdfe612a Support for minimum bandwidth rules in tunnelled networks\n572185e1fd Delete the PB level registers when deleting the duplicated PB\nbe0776d647 Add sanity check for keepalived gar on SIGHUP support\nb040e6139e [Fullstack] use common assert for ncat connection\n3d575f8bd0 Add an env variable \"PROCESS_TAG\" in ``ProcessManager``\n08e485d0cd Ignore python warnings in the neutron-status functional tests\n370d8bcea3 [Fullstack] Wait 10 seconds to ensure that MAC address is configured\n4edf94041f Remove note about migration from lib/neutron-legacy to lib/neutron\n0475a3e3f3 Fix typos in routed networks doc file\ndad23fdcdb Strictly delete arp_spoofing_protection flows\nfaa4803302 dhcp: fix issue when network is already removed\nbe0996c308 segment: enable multisegments support for host\n7c449f1833 Enable ML2/OVS backend in the -ovs- grenade jobs explicitly\n9401da6c7c Handle if \"TRUNK_SKELETON\" is not loaded\nb5fdad3cfb Pin OVS_BRANCH in ipv6 ovn job\n14d0ff0317 Add ``devlink.get_port`` method for devlink ports\n0d315d2467 Update .pylintrc\n86badcfe2d Fix some pylint indentation warnings\n4e75c69ab6 Add extra logs when checking if FIP GW port has been created\ncbd03c7ea5 Wait until local OVS DB updates Bridge \"datapath_id\"\n1d46b19951 Pin OVS_BRANCH to master again\n4cc611d319 Fix handling the restart of ovn-controllers\n44ad4bb517 Remove deprecated \"list_moved_globals\" script\n8bcaf1632f Limit tox version to \u003c4\nc2fb0b16b6 Rename setup-mtu-t1 argument because it is misleading\n135daae62d Use same ovs/ovn versions in local test setups\naaae079883 Fix homedir permissions\n405414772e Fix tox.ini for tox4\nca73925f42 Pin OVS_BRANCH to working commit\nf7c0f466be Fix text in OVN RevisionConflict exception to match code\nc7cd387796 Amend documentation for fullstack tests\ndfe69472a8 Check if port exists in ``update_port_virtual_type`` method\n2fd17aa876 [OVN] Initialize the mech driver variable in ``OVNL3RouterPlugin``\n06fad386ba Fix firewall_driver examples in docs\n74224e79e0 Gracefully restart dnsmasq to not break tcp DNS\nf22aa5dfdd Discard port with ofport -1 in _get_ofport_moves\n81980146cb Add and delete multiple ip addresses in one priv call\n92efd8e45b Dont raise RouterInterfaceNotFound on overlap check router ports\n49eab7d308 Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init\n234c5a376b [sqlalchemy-20] Add missing DB context decorator\n88b7db0e0c Mock ls_get in OVN metadata delete network unit test\na8adb5aa9c [OVN] Implement abstract \"post_connect\" in \"MetadataAgentOvnSbIdl\"\n81330921d8 Fix some pylint indentation warnings\ncca0b3a026 Fix some pylint indentation warnings\nfd423f7b1b Reset the MeterIDGenerator singleton after each test\n4416706931 Use neutron-lib constants for OVS table numbers\n26c55efe34 Do not emit ssl warnings in the test_ssl_connection test\nf7dd7790f5 Fix bulk create without mac\n895f818cb3 Do not mock the hash method for testing \"get_vlan_device_name\"\n5f819c0767 Update neutron-tox-override-jobs template\nc813b658d0 Fix duplicated routes exceptions\nf629b77d3c Fix behaviour of enable/disable in OVN network log\n5dad500481 Remove deprecation warning of \"ovs-ofctl\" usage\n2272b9be95 Add address scope to the OVN LSP port registers\n499ae95881 Bump pecan to fix #1829042\n7a7448ead4 Configure placement credentials for segments plugin\n6d8ada0ac9 [S-RBAC] Allow admin user to do all API requests by default\n4d1a7bd0bc neutron-dynamic-routing now works with OVN\n3b1fd52b41 Exclude neutron-dynamic-routing from abandon script\ne1d70f9cb0 Disable \"test_live_migration_with_trunk\" execution from dvr job\neeff5b3c81 db: add segment_index to the unique constraint\n67bd2badc9 Fix missing [designate] options\n631ad0ce6b ovn: use dicts instead of sets of tuples\nae845d572c ovn: inline ensure_all_networks_provisioned method\n686698284b Update tap ip in metadata agent when metadata port ip updated\n1a9ae924fd Add VXLAN support reference in the OVN installation guide\n3093aaab13 ovn: first tear down old metadata namespaces, then deploy new\n0ef4f98825 Remove policy rule for get_network:router:external\n2081910d6d Always create a \"router_extra_attributes\" register per router\n5b64ac9013 Use the neutron-lib constant for PACKET_RATE_LIMIT table number (2)\n9a483f02be Make MechanismDriverContext plugin_context public\n33de608f04 Avoid race condition when deleting trunk bridges\nc5640eee11 Use the neutron-lib constant for PACKET_RATE_LIMIT table number\nbf44e70db6 OVN: Add support for DHCP option \"domain-search\" for IPv4\n25d4e3862b Code move for metadata haproxy\nc3ebefa5f7 Refactor for meter ID Generator\nff66ca9275 Execute functional test \"test_get_all_devices\" alone\n51827d8e78 Remove \"test_live_migration_with_trunk\" execution\na2090e4cbc Doc: Capitalize keystone domain names\nc5ee9f3495 Load the required configuration options in the UT classes\nbb2d706107 [doc] Metadata service supports IPv6 networks\nb7d04d5d92 Code move for metadata signature function\nc6a83988ea Replace Linux Bridge with Open vSwitch in installation docs\n81004125c4 [OVN] Prevent maintenance task fail resource stdattr does not exist\n9e2a0ac058 Use --exclude-regex in tox.ini\n609ebd9504 test: fix docstring validation function\n562e9704f8 Update documentation link for openSUSE index.\nb1714a2b9d Fix some pylint indentation warnings\n55b16d7b7c Fix some pylint indentation warnings\n3afa1b1471 Fix some pylint indentation warnings\n3702575bfe Fix some pylint indentation warnings\nf77c7c9584 Fix some pylint indentation warnings\n291eabb0b6 Fix some pylint indentation warnings\nd1c2d2c4fe dhcp: support multiple segmentations per network\n01b11df1b5 Add tempest job which will run on the oldest Ubuntu release\n0da328fbee Add \"tempest-integrated-networking\" job to the check and gate queues\n1abb77d7a6 Check subnet overlapping after add router interface\n304394ed77 Revert \"Temporary make rally job non voting\"\nc871135cdd Remove xenapi from neutron ml2 config opts\n8fcf00a36d Disable in-band management for bridges before setting up controllers\nc33b47edc7 Do not keep gateway port when notifying for router update\n7c1a894ce5 Nit: network_update in ovs_neutron_agent has a bad LOG\n963f8e30d8 CI: Add tick-tick grenade jobs to check queue\nda335193ef Promote neutron-ovn-rally-task job to be voting in check queue\n8d5985ba6a [Docs] Remove python version from the ci jobs summary\n99339a9704 Clean \"ovn ipv6 only\" jobs templates\nb7b8f7c571 [OVN] Avoid deadlock when cleaning hash ring nodes\ncb332acb29 dhcp: add/use cleanup stale devices API\n1d8e3b79db dhcp: make device manager to clean only not used devices\n4551718bad Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\nee7bfca937 Remove unnecessary bullet points, unify formatting.\n7a5a61616a Remove reference to \u0027all-plugin\u0027 tox environment\n72dd768a12 Revert \"Since OVN 20.06, config is stored in \"Chassis.other_config\"\"\n628442aed7 Spread OVN metadata agent heartbeat response in time\nca2e593177 Change grenade job names suffix to \"skip-level\"\n240f2c6aeb Split Hash Ring probing from the maintenance task\n1920f2a4e3 Add new oslo master CI jobs to the periodic queue\n7b8b797dea Load the required configuration options in the UT classes\n4858315286 Fix misplaced comparison constant warnings\n3d0f1ef4fa Remove some unnecessary list comprehensions\n70a2029ca1 [OVN] Set the default OVN metadata worker number to 0\n34b81ae9b7 Remove OVN gap minimum bandwidth support\n229d798a45 [OVN] Allow only one physical network per bridge\nf43891bf86 [OVN] Allow to execute ``MetadataProxyHandler`` in a local thread\n50eee19723 ovn: Use ovsdb-client to create neutron_pg_drop\n536498a29a Since OVN 20.06, config is stored in \"Chassis.other_config\"\n0c9e7d4e18 Update grenade skip level jobs for new release\nde18dd2c2e [ovn migration] Use ecsda ssh key instead of rsa\nd7b44f7218 Allow to pass EUI64 IP address as fixed ip for the port\na3e68e8f76 Fix ipam_pluggable_backend unit tests module\na8a43c8045 Execute \"IpMonitorTestCase\" tests always inside a namespace\nf8fa909444 Since OVN 20.06, config is stored in \"Chassis.other_config\"\n7b15ad9433 Bump minimum OVN version to v21.06.0\ne4cc40f114 fix: Fix url of Floodlight\n21491efd9f Port provisioning should retry only for VM ports\n178ee6fd3d [Trunk] Update the trunk status with the parent status\n0b2844b8a8 Add .pylintrc to irrelevant-files for non pep8 jobs\n72493d1472 Add missing comma in .pylintrc\n3f20cabc4f Revert \"Move the \"ovn_hash_ring\" clean up to maintenance worker\"\n56f89131dd docs: Remove errant indentation, mark up literals\n2f0919f7df Accept a port deletion with missing port binding information\n1fe4edab80 Open Antelope DB branch\n41a7479d0f Handle several dhcp agents for metadata over ipv6\n819a1bb3e6 Move the \"ovn_hash_ring\" clean up to maintenance worker\ne9b5a27001 Use \"OVNMechDriver\" instance in \"TestOvn[Nb|Sb]IdlNotifyHandler\"\n9eddb132b2 Fix openstack-tox-py39-with-oslo-master job\na765b7acbb [sqlalchemy-20] Add py39 oslo master to experimental\nea11928ba0 Neutron needs oslo.config \u003e\u003d 9\n9a1b5d1e51 Imported Translations from Zanata\n46bc1e7f13 Switch to 2023.1 Python3 unit tests and generic template name\n4914d7fab2 Update master for stable/zed\n639257c77b Make path for tcpdump release-specific\ncca48be73d Imported Translations from Zanata\nf7578e51f2 Collect ovn/ovs logs in singlenode jobs\n9655466763 [OVN] Rate limit the \"Disallow caching\" log from hash ring\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Iefe51ff7ca7e451d4149ad8886896e990d7066e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b783e08c84dfe7ecf92a1d97778659e598fd270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b783e08c84dfe7ecf92a1d97778659e598fd270"}]},"branch":"refs/heads/master"},"4a550e737c386f73523158b148ccc7b238ef033a":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:18:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/875446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/875446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/875446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/875446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/875446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/875446/2"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:03:00.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 14:18:29.000000000","tz":0},"subject":"Release neutron RC1 for Antelope","message":"Release neutron RC1 for Antelope\n\nThis proposed a RC1 release for neutron to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 21.0.0.0rc1..b30d7a2d03\naa40aef70f Mark fullstack TestMultiSegs.test_multi_segs_network as unstable\nf991fc2fbe [ovn][ipv6] Skip test_update_delete_extra_route\n51d449babf Move tox-py311 job override into job-templates.yaml\nc61ce3447e Format correctly (dialect\u003dmac_unix_expanded) the MAC addresses\n502c0253de [ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class\ne9ce025a69 [dhcp agent] Fix clean_devices\nd052da85cf Mark ``test_port_creation_and_deletion`` as unstable\n5e15c4a4d4 Allow other kwargs in ``DhcpBase.disable``\n5dac5f8c6c [ovn-ipv6] Temporary exclude test_hotplug_nic\n67a232fe71 Remove tripleo related jobs from neutron queues\n595c387efb [UT] Mock the retry decorator when creating duplicated NDPProxy\nf9880c2013 Increase tox-py311 timeout to 3600 seconds\n24152f2019 Add debugging exception messages in ``test_ovs_and_ovs_events``\nfce516e346 [OVN] Bump the port revision number in trunk driver\nd83e243c6d [OVN] Add missing LSP device_owner info in trunk driver\n76f3b75ad6 [OVN] Skip ports with VNIC type direct and no port binding\n6df75ca696 Deprecate allow_stateless_action_supported config option\n5250598c80 Improve scheduling L3/DHCP agents, missing lower binding indexes\n8f7cc690a9 Remove ``L3Scheduler._bind_routers`` method\n47fef55e25 Add a method to retrieve and register \"rpc_workers\" config knob\n359030f741 Do not log full network on reload_allocations\n4e71675ddb rbacs: add functionl test that asserts behavior on networks/users\n5c697b8d60 allow manila ports to do multiple port binding for ML2\nc15313ea21 Removed duplicated job call in experimental queue\nc93c31852b Adapt deploy_rootwrap filters path for tox4\n61b2917a3e Filter subnets by \"enable_dhcp\" flag using the correct type\n8be42f6d72 Enable qos extension_driver in ovn_db_sync\n3c2af924d9 Add CI jobs using SQLAlchemy master branch\nabfa8940b1 [OVN] Fix the OVN Agent config file location\n54eff20a72 [OVN] New OVN Neutron Agent extension: QoS for HWOL\n07fd743dc9 Remove extra call to config_logging() for neutron-server\ne98a268de4 Propose replacement of ORM from_self()\n1c27d8b5b4 [Secure RBAC] Fix policy to get flavors\n17be2505d1 [Secure RBAC] Allow network owner to delete ports plugged to the network\n5d505e7b7c [Secure RBAC] Add shared_security_groups rule\n221cf9497f [Secure RBAC] Add shared_qos_policy rule\nf7e31b4c05 [OVN] Allow logging all traffic related to an ACL\n267ad86291 Add tox Python3.11 job to the testing queues\n83da1e6d79 Improve message for subnet gateway out of host IP addresses range\n7cb481a3dc Allow disable stateful security group extension on older OVN\n48c3621292 Imported Translations from Zanata\n3727103ce7 Fix get_link_devices() with index argument\n67e616b238 Never raise an exception in notify()\nd0c7bb653a [OVN] Implementation of OVN Neutron Agent\nc609084b59 Add doc note on nf_conntrack module requirement\n0ae432e717 Fullstack: Wait placement process fixtrue to really stop\n910574a5cc Use SLQAlchemy ORM \"relationship\" instead of \"relation\"\n1d22cfa99a Change flag check order in wait_until_address_ready()\ned68ba4a4c Do not ignore attributes in bulk port create\nd73f75c551 [API] Add API extension and definition for default SG rules\n1a3bdff18a Honor debug mode in keepalived-state-change script logs\nec438577bd Use dhcpcd client in the tempest slow jobs\nf235ce8791 Add new debugging logs for ``ARPSpoofTestCase``\n78c3d52bb4 [OVN] Introduce the new OVN Neutron Agent\nfeb9f2b21c ovn-migration: Stop neutron server while running db sync\n2495ad2856 Avoid failure of get_device_details if net from cache is empty\nb6750fb2b8 [OVN] Add vnic_type and binding profile capabilities to LSP info\na3bf74baa5 Ignore reno artefacts (RELEASENOTES.rst and reno.cache)\n1d84aaf877 Bump the minimum pyroute2 version to 0.7.3\n463c3df4cf [ovn] Do not create empty default route when empty gateway_ip\n70a86637e7 Allow shared resources between physical and tunnelled networks\nf28e416599 Add 3 secs to wait for keepalived state change\nd7f68a0ce7 Set proxy_delay to one when using proxy ARP\nd0e64c6617 Add auto-allocated-topology delete example to docs\n3354b43d5a Increase fullstack job\u0027s timeout\ne2097157bb Allow multiple IPv6 ports on router from same network ml2/ovs+dvr\nde0918cea3 Add debug logs to OVSInterfaceDriver.plug_new\n7751ac32a7 Add info on segments extension to contrib guide\ned048638f4 Add text to WSGI config guide for debugging\nc362016620 Ensure that MAC address of the device is set correctly\n3e1e2d63b3 Enable rpc notifications only when rpc_workers \u003e\u003d 1\n1c89db495c [OVN][FT] Also update nb_cfg along with nb_cfg_timestamp\nd59bf5ff6c Temporary update envlist for tox4\n25ec6e7e4f Set ip_nonlocal_bind to 1 for HA routers and DVR snat\n07fd0e101b Revert \"Documentation for adding extra port conflict fix\"\n6501ebc6ca Documentation for adding extra port conflict fix\nf759915ab0 Fix default value for MTUs, when not provided\n02030f037a ovs-dpdk support in neutron-sanity-check\n9f9478071e Bump Cirros version used in CI jobs to 0.6.1\nbee07defac Track all interfaces in Keepalived\n26da863ca6 Remove deprecated opt keepalived_use_no_track\n4fa85b9a99 Fix incorrect mysql examples\nedf48e46a1 Improve agent provision performance for large networks\n5e19eaadbb [OVN] Fix availability zones changes check\n3ebdfe612a Support for minimum bandwidth rules in tunnelled networks\n572185e1fd Delete the PB level registers when deleting the duplicated PB\nbe0776d647 Add sanity check for keepalived gar on SIGHUP support\nb040e6139e [Fullstack] use common assert for ncat connection\n3d575f8bd0 Add an env variable \"PROCESS_TAG\" in ``ProcessManager``\n08e485d0cd Ignore python warnings in the neutron-status functional tests\n370d8bcea3 [Fullstack] Wait 10 seconds to ensure that MAC address is configured\n4edf94041f Remove note about migration from lib/neutron-legacy to lib/neutron\n0475a3e3f3 Fix typos in routed networks doc file\ndad23fdcdb Strictly delete arp_spoofing_protection flows\nfaa4803302 dhcp: fix issue when network is already removed\nbe0996c308 segment: enable multisegments support for host\n7c449f1833 Enable ML2/OVS backend in the -ovs- grenade jobs explicitly\n9401da6c7c Handle if \"TRUNK_SKELETON\" is not loaded\nb5fdad3cfb Pin OVS_BRANCH in ipv6 ovn job\n14d0ff0317 Add ``devlink.get_port`` method for devlink ports\n0d315d2467 Update .pylintrc\n86badcfe2d Fix some pylint indentation warnings\n4e75c69ab6 Add extra logs when checking if FIP GW port has been created\ncbd03c7ea5 Wait until local OVS DB updates Bridge \"datapath_id\"\n1d46b19951 Pin OVS_BRANCH to master again\n4cc611d319 Fix handling the restart of ovn-controllers\n44ad4bb517 Remove deprecated \"list_moved_globals\" script\n8bcaf1632f Limit tox version to \u003c4\nc2fb0b16b6 Rename setup-mtu-t1 argument because it is misleading\n135daae62d Use same ovs/ovn versions in local test setups\naaae079883 Fix homedir permissions\n405414772e Fix tox.ini for tox4\nca73925f42 Pin OVS_BRANCH to working commit\nf7c0f466be Fix text in OVN RevisionConflict exception to match code\nc7cd387796 Amend documentation for fullstack tests\ndfe69472a8 Check if port exists in ``update_port_virtual_type`` method\n2fd17aa876 [OVN] Initialize the mech driver variable in ``OVNL3RouterPlugin``\n06fad386ba Fix firewall_driver examples in docs\n74224e79e0 Gracefully restart dnsmasq to not break tcp DNS\nf22aa5dfdd Discard port with ofport -1 in _get_ofport_moves\n81980146cb Add and delete multiple ip addresses in one priv call\n92efd8e45b Dont raise RouterInterfaceNotFound on overlap check router ports\n49eab7d308 Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init\n234c5a376b [sqlalchemy-20] Add missing DB context decorator\n88b7db0e0c Mock ls_get in OVN metadata delete network unit test\na8adb5aa9c [OVN] Implement abstract \"post_connect\" in \"MetadataAgentOvnSbIdl\"\n81330921d8 Fix some pylint indentation warnings\ncca0b3a026 Fix some pylint indentation warnings\nfd423f7b1b Reset the MeterIDGenerator singleton after each test\n4416706931 Use neutron-lib constants for OVS table numbers\n26c55efe34 Do not emit ssl warnings in the test_ssl_connection test\nf7dd7790f5 Fix bulk create without mac\n895f818cb3 Do not mock the hash method for testing \"get_vlan_device_name\"\n5f819c0767 Update neutron-tox-override-jobs template\nc813b658d0 Fix duplicated routes exceptions\nf629b77d3c Fix behaviour of enable/disable in OVN network log\n5dad500481 Remove deprecation warning of \"ovs-ofctl\" usage\n2272b9be95 Add address scope to the OVN LSP port registers\n499ae95881 Bump pecan to fix #1829042\n7a7448ead4 Configure placement credentials for segments plugin\n6d8ada0ac9 [S-RBAC] Allow admin user to do all API requests by default\n4d1a7bd0bc neutron-dynamic-routing now works with OVN\n3b1fd52b41 Exclude neutron-dynamic-routing from abandon script\ne1d70f9cb0 Disable \"test_live_migration_with_trunk\" execution from dvr job\neeff5b3c81 db: add segment_index to the unique constraint\n67bd2badc9 Fix missing [designate] options\n631ad0ce6b ovn: use dicts instead of sets of tuples\nae845d572c ovn: inline ensure_all_networks_provisioned method\n686698284b Update tap ip in metadata agent when metadata port ip updated\n1a9ae924fd Add VXLAN support reference in the OVN installation guide\n3093aaab13 ovn: first tear down old metadata namespaces, then deploy new\n0ef4f98825 Remove policy rule for get_network:router:external\n2081910d6d Always create a \"router_extra_attributes\" register per router\n5b64ac9013 Use the neutron-lib constant for PACKET_RATE_LIMIT table number (2)\n9a483f02be Make MechanismDriverContext plugin_context public\n33de608f04 Avoid race condition when deleting trunk bridges\nc5640eee11 Use the neutron-lib constant for PACKET_RATE_LIMIT table number\nbf44e70db6 OVN: Add support for DHCP option \"domain-search\" for IPv4\n25d4e3862b Code move for metadata haproxy\nc3ebefa5f7 Refactor for meter ID Generator\nff66ca9275 Execute functional test \"test_get_all_devices\" alone\n51827d8e78 Remove \"test_live_migration_with_trunk\" execution\na2090e4cbc Doc: Capitalize keystone domain names\nc5ee9f3495 Load the required configuration options in the UT classes\nbb2d706107 [doc] Metadata service supports IPv6 networks\nb7d04d5d92 Code move for metadata signature function\nc6a83988ea Replace Linux Bridge with Open vSwitch in installation docs\n81004125c4 [OVN] Prevent maintenance task fail resource stdattr does not exist\n9e2a0ac058 Use --exclude-regex in tox.ini\n609ebd9504 test: fix docstring validation function\n562e9704f8 Update documentation link for openSUSE index.\nb1714a2b9d Fix some pylint indentation warnings\n55b16d7b7c Fix some pylint indentation warnings\n3afa1b1471 Fix some pylint indentation warnings\n3702575bfe Fix some pylint indentation warnings\nf77c7c9584 Fix some pylint indentation warnings\n291eabb0b6 Fix some pylint indentation warnings\nd1c2d2c4fe dhcp: support multiple segmentations per network\n01b11df1b5 Add tempest job which will run on the oldest Ubuntu release\n0da328fbee Add \"tempest-integrated-networking\" job to the check and gate queues\n1abb77d7a6 Check subnet overlapping after add router interface\n304394ed77 Revert \"Temporary make rally job non voting\"\nc871135cdd Remove xenapi from neutron ml2 config opts\n8fcf00a36d Disable in-band management for bridges before setting up controllers\nc33b47edc7 Do not keep gateway port when notifying for router update\n7c1a894ce5 Nit: network_update in ovs_neutron_agent has a bad LOG\n963f8e30d8 CI: Add tick-tick grenade jobs to check queue\nda335193ef Promote neutron-ovn-rally-task job to be voting in check queue\n8d5985ba6a [Docs] Remove python version from the ci jobs summary\n99339a9704 Clean \"ovn ipv6 only\" jobs templates\nb7b8f7c571 [OVN] Avoid deadlock when cleaning hash ring nodes\ncb332acb29 dhcp: add/use cleanup stale devices API\n1d8e3b79db dhcp: make device manager to clean only not used devices\n4551718bad Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\nee7bfca937 Remove unnecessary bullet points, unify formatting.\n7a5a61616a Remove reference to \u0027all-plugin\u0027 tox environment\n72dd768a12 Revert \"Since OVN 20.06, config is stored in \"Chassis.other_config\"\"\n628442aed7 Spread OVN metadata agent heartbeat response in time\nca2e593177 Change grenade job names suffix to \"skip-level\"\n240f2c6aeb Split Hash Ring probing from the maintenance task\n1920f2a4e3 Add new oslo master CI jobs to the periodic queue\n7b8b797dea Load the required configuration options in the UT classes\n4858315286 Fix misplaced comparison constant warnings\n3d0f1ef4fa Remove some unnecessary list comprehensions\n70a2029ca1 [OVN] Set the default OVN metadata worker number to 0\n34b81ae9b7 Remove OVN gap minimum bandwidth support\n229d798a45 [OVN] Allow only one physical network per bridge\nf43891bf86 [OVN] Allow to execute ``MetadataProxyHandler`` in a local thread\n50eee19723 ovn: Use ovsdb-client to create neutron_pg_drop\n536498a29a Since OVN 20.06, config is stored in \"Chassis.other_config\"\n0c9e7d4e18 Update grenade skip level jobs for new release\nde18dd2c2e [ovn migration] Use ecsda ssh key instead of rsa\nd7b44f7218 Allow to pass EUI64 IP address as fixed ip for the port\na3e68e8f76 Fix ipam_pluggable_backend unit tests module\na8a43c8045 Execute \"IpMonitorTestCase\" tests always inside a namespace\nf8fa909444 Since OVN 20.06, config is stored in \"Chassis.other_config\"\n7b15ad9433 Bump minimum OVN version to v21.06.0\ne4cc40f114 fix: Fix url of Floodlight\n21491efd9f Port provisioning should retry only for VM ports\n178ee6fd3d [Trunk] Update the trunk status with the parent status\n0b2844b8a8 Add .pylintrc to irrelevant-files for non pep8 jobs\n72493d1472 Add missing comma in .pylintrc\n3f20cabc4f Revert \"Move the \"ovn_hash_ring\" clean up to maintenance worker\"\n56f89131dd docs: Remove errant indentation, mark up literals\n2f0919f7df Accept a port deletion with missing port binding information\n1fe4edab80 Open Antelope DB branch\n41a7479d0f Handle several dhcp agents for metadata over ipv6\n819a1bb3e6 Move the \"ovn_hash_ring\" clean up to maintenance worker\ne9b5a27001 Use \"OVNMechDriver\" instance in \"TestOvn[Nb|Sb]IdlNotifyHandler\"\n9eddb132b2 Fix openstack-tox-py39-with-oslo-master job\na765b7acbb [sqlalchemy-20] Add py39 oslo master to experimental\nea11928ba0 Neutron needs oslo.config \u003e\u003d 9\n9a1b5d1e51 Imported Translations from Zanata\n46bc1e7f13 Switch to 2023.1 Python3 unit tests and generic template name\n4914d7fab2 Update master for stable/zed\n639257c77b Make path for tcpdump release-specific\ncca48be73d Imported Translations from Zanata\nf7578e51f2 Collect ovn/ovs logs in singlenode jobs\n9655466763 [OVN] Rate limit the \"Disallow caching\" log from hash ring\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Iefe51ff7ca7e451d4149ad8886896e990d7066e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a550e737c386f73523158b148ccc7b238ef033a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a550e737c386f73523158b148ccc7b238ef033a"}]},"branch":"refs/heads/master"},"66b571ee5283e406ccb31b7c48235fbdda20b041":{"kind":"REWORK","_number":3,"created":"2023-03-02 08:42:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/875446/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/875446/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/875446/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/875446/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/875446/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/875446/3"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:03:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-02 08:42:01.000000000","tz":0},"subject":"Release neutron RC1 for Antelope","message":"Release neutron RC1 for Antelope\n\nThis proposed a RC1 release for neutron to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 21.0.0.0rc1..41846a6e8b\n41846a6e8b [sqlalchemy-20] The Connection.connect() method is considered legacy\n19482a049b Add Slawomir Kaplonski as Lieutenant in API, Client and Testing areas\n51005388b8 Make retrieval of port mac column safe\n6cac935824 Update lieutenants for ovn-octavia-provider\n948c9e02e3 [S-RBAC] Add release note about full support for new policies\n289ae97c1c Add missing osprofiler options\nacb809eea4 [OVN] Ensure traffic for provider vlan networks is not tunneled\n7f46ea9b6d Remove unused rbac rule\nc2a02d3c69 Remove two duplicated unit tests\ndbfc18d1fa Apply Ironic\u0027s server-ip-address as TFTP next-server\nbdf0c8bdad Move register_common_config_options call to DietTestCase\ne7c92d92e4 [sqlalchemy-20] Use the .begin() method of Engine\n827fbd01c3 Normalise format of OVN agent heartbeat timestamp\ne68e4162ce Prevent router_ha_interface port from being removed via API\naa40aef70f Mark fullstack TestMultiSegs.test_multi_segs_network as unstable\nf991fc2fbe [ovn][ipv6] Skip test_update_delete_extra_route\n51d449babf Move tox-py311 job override into job-templates.yaml\nc61ce3447e Format correctly (dialect\u003dmac_unix_expanded) the MAC addresses\n502c0253de [ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class\ne9ce025a69 [dhcp agent] Fix clean_devices\nd052da85cf Mark ``test_port_creation_and_deletion`` as unstable\n5e15c4a4d4 Allow other kwargs in ``DhcpBase.disable``\n5dac5f8c6c [ovn-ipv6] Temporary exclude test_hotplug_nic\n67a232fe71 Remove tripleo related jobs from neutron queues\n595c387efb [UT] Mock the retry decorator when creating duplicated NDPProxy\nf9880c2013 Increase tox-py311 timeout to 3600 seconds\n24152f2019 Add debugging exception messages in ``test_ovs_and_ovs_events``\nfce516e346 [OVN] Bump the port revision number in trunk driver\nd83e243c6d [OVN] Add missing LSP device_owner info in trunk driver\n76f3b75ad6 [OVN] Skip ports with VNIC type direct and no port binding\n6df75ca696 Deprecate allow_stateless_action_supported config option\n5250598c80 Improve scheduling L3/DHCP agents, missing lower binding indexes\n8f7cc690a9 Remove ``L3Scheduler._bind_routers`` method\n47fef55e25 Add a method to retrieve and register \"rpc_workers\" config knob\n359030f741 Do not log full network on reload_allocations\n4e71675ddb rbacs: add functionl test that asserts behavior on networks/users\n5c697b8d60 allow manila ports to do multiple port binding for ML2\nc15313ea21 Removed duplicated job call in experimental queue\nc93c31852b Adapt deploy_rootwrap filters path for tox4\n61b2917a3e Filter subnets by \"enable_dhcp\" flag using the correct type\n8be42f6d72 Enable qos extension_driver in ovn_db_sync\n3c2af924d9 Add CI jobs using SQLAlchemy master branch\nabfa8940b1 [OVN] Fix the OVN Agent config file location\n54eff20a72 [OVN] New OVN Neutron Agent extension: QoS for HWOL\n07fd743dc9 Remove extra call to config_logging() for neutron-server\ne98a268de4 Propose replacement of ORM from_self()\n1c27d8b5b4 [Secure RBAC] Fix policy to get flavors\n17be2505d1 [Secure RBAC] Allow network owner to delete ports plugged to the network\n5d505e7b7c [Secure RBAC] Add shared_security_groups rule\n221cf9497f [Secure RBAC] Add shared_qos_policy rule\nf7e31b4c05 [OVN] Allow logging all traffic related to an ACL\n267ad86291 Add tox Python3.11 job to the testing queues\n83da1e6d79 Improve message for subnet gateway out of host IP addresses range\n7cb481a3dc Allow disable stateful security group extension on older OVN\n48c3621292 Imported Translations from Zanata\n3727103ce7 Fix get_link_devices() with index argument\n67e616b238 Never raise an exception in notify()\nd0c7bb653a [OVN] Implementation of OVN Neutron Agent\nc609084b59 Add doc note on nf_conntrack module requirement\n0ae432e717 Fullstack: Wait placement process fixtrue to really stop\n910574a5cc Use SLQAlchemy ORM \"relationship\" instead of \"relation\"\n1d22cfa99a Change flag check order in wait_until_address_ready()\ned68ba4a4c Do not ignore attributes in bulk port create\nd73f75c551 [API] Add API extension and definition for default SG rules\n1a3bdff18a Honor debug mode in keepalived-state-change script logs\nec438577bd Use dhcpcd client in the tempest slow jobs\nf235ce8791 Add new debugging logs for ``ARPSpoofTestCase``\n78c3d52bb4 [OVN] Introduce the new OVN Neutron Agent\nfeb9f2b21c ovn-migration: Stop neutron server while running db sync\n2495ad2856 Avoid failure of get_device_details if net from cache is empty\nb6750fb2b8 [OVN] Add vnic_type and binding profile capabilities to LSP info\na3bf74baa5 Ignore reno artefacts (RELEASENOTES.rst and reno.cache)\n1d84aaf877 Bump the minimum pyroute2 version to 0.7.3\n463c3df4cf [ovn] Do not create empty default route when empty gateway_ip\n70a86637e7 Allow shared resources between physical and tunnelled networks\nf28e416599 Add 3 secs to wait for keepalived state change\nd7f68a0ce7 Set proxy_delay to one when using proxy ARP\nd0e64c6617 Add auto-allocated-topology delete example to docs\n3354b43d5a Increase fullstack job\u0027s timeout\ne2097157bb Allow multiple IPv6 ports on router from same network ml2/ovs+dvr\nde0918cea3 Add debug logs to OVSInterfaceDriver.plug_new\n7751ac32a7 Add info on segments extension to contrib guide\ned048638f4 Add text to WSGI config guide for debugging\nc362016620 Ensure that MAC address of the device is set correctly\n3e1e2d63b3 Enable rpc notifications only when rpc_workers \u003e\u003d 1\n1c89db495c [OVN][FT] Also update nb_cfg along with nb_cfg_timestamp\nd59bf5ff6c Temporary update envlist for tox4\n25ec6e7e4f Set ip_nonlocal_bind to 1 for HA routers and DVR snat\n07fd0e101b Revert \"Documentation for adding extra port conflict fix\"\n6501ebc6ca Documentation for adding extra port conflict fix\nf759915ab0 Fix default value for MTUs, when not provided\n02030f037a ovs-dpdk support in neutron-sanity-check\n9f9478071e Bump Cirros version used in CI jobs to 0.6.1\nbee07defac Track all interfaces in Keepalived\n26da863ca6 Remove deprecated opt keepalived_use_no_track\n4fa85b9a99 Fix incorrect mysql examples\nedf48e46a1 Improve agent provision performance for large networks\n5e19eaadbb [OVN] Fix availability zones changes check\n3ebdfe612a Support for minimum bandwidth rules in tunnelled networks\n572185e1fd Delete the PB level registers when deleting the duplicated PB\nbe0776d647 Add sanity check for keepalived gar on SIGHUP support\nb040e6139e [Fullstack] use common assert for ncat connection\n3d575f8bd0 Add an env variable \"PROCESS_TAG\" in ``ProcessManager``\n08e485d0cd Ignore python warnings in the neutron-status functional tests\n370d8bcea3 [Fullstack] Wait 10 seconds to ensure that MAC address is configured\n4edf94041f Remove note about migration from lib/neutron-legacy to lib/neutron\n0475a3e3f3 Fix typos in routed networks doc file\ndad23fdcdb Strictly delete arp_spoofing_protection flows\nfaa4803302 dhcp: fix issue when network is already removed\nbe0996c308 segment: enable multisegments support for host\n7c449f1833 Enable ML2/OVS backend in the -ovs- grenade jobs explicitly\n9401da6c7c Handle if \"TRUNK_SKELETON\" is not loaded\nb5fdad3cfb Pin OVS_BRANCH in ipv6 ovn job\n14d0ff0317 Add ``devlink.get_port`` method for devlink ports\n0d315d2467 Update .pylintrc\n86badcfe2d Fix some pylint indentation warnings\n4e75c69ab6 Add extra logs when checking if FIP GW port has been created\ncbd03c7ea5 Wait until local OVS DB updates Bridge \"datapath_id\"\n1d46b19951 Pin OVS_BRANCH to master again\n4cc611d319 Fix handling the restart of ovn-controllers\n44ad4bb517 Remove deprecated \"list_moved_globals\" script\n8bcaf1632f Limit tox version to \u003c4\nc2fb0b16b6 Rename setup-mtu-t1 argument because it is misleading\n135daae62d Use same ovs/ovn versions in local test setups\naaae079883 Fix homedir permissions\n405414772e Fix tox.ini for tox4\nca73925f42 Pin OVS_BRANCH to working commit\nf7c0f466be Fix text in OVN RevisionConflict exception to match code\nc7cd387796 Amend documentation for fullstack tests\ndfe69472a8 Check if port exists in ``update_port_virtual_type`` method\n2fd17aa876 [OVN] Initialize the mech driver variable in ``OVNL3RouterPlugin``\n06fad386ba Fix firewall_driver examples in docs\n74224e79e0 Gracefully restart dnsmasq to not break tcp DNS\nf22aa5dfdd Discard port with ofport -1 in _get_ofport_moves\n81980146cb Add and delete multiple ip addresses in one priv call\n92efd8e45b Dont raise RouterInterfaceNotFound on overlap check router ports\n49eab7d308 Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init\n234c5a376b [sqlalchemy-20] Add missing DB context decorator\n88b7db0e0c Mock ls_get in OVN metadata delete network unit test\na8adb5aa9c [OVN] Implement abstract \"post_connect\" in \"MetadataAgentOvnSbIdl\"\n81330921d8 Fix some pylint indentation warnings\ncca0b3a026 Fix some pylint indentation warnings\nfd423f7b1b Reset the MeterIDGenerator singleton after each test\n4416706931 Use neutron-lib constants for OVS table numbers\n26c55efe34 Do not emit ssl warnings in the test_ssl_connection test\nf7dd7790f5 Fix bulk create without mac\n895f818cb3 Do not mock the hash method for testing \"get_vlan_device_name\"\n5f819c0767 Update neutron-tox-override-jobs template\nc813b658d0 Fix duplicated routes exceptions\nf629b77d3c Fix behaviour of enable/disable in OVN network log\n5dad500481 Remove deprecation warning of \"ovs-ofctl\" usage\n2272b9be95 Add address scope to the OVN LSP port registers\n499ae95881 Bump pecan to fix #1829042\n7a7448ead4 Configure placement credentials for segments plugin\n6d8ada0ac9 [S-RBAC] Allow admin user to do all API requests by default\n4d1a7bd0bc neutron-dynamic-routing now works with OVN\n3b1fd52b41 Exclude neutron-dynamic-routing from abandon script\ne1d70f9cb0 Disable \"test_live_migration_with_trunk\" execution from dvr job\neeff5b3c81 db: add segment_index to the unique constraint\n67bd2badc9 Fix missing [designate] options\n631ad0ce6b ovn: use dicts instead of sets of tuples\nae845d572c ovn: inline ensure_all_networks_provisioned method\n686698284b Update tap ip in metadata agent when metadata port ip updated\n1a9ae924fd Add VXLAN support reference in the OVN installation guide\n3093aaab13 ovn: first tear down old metadata namespaces, then deploy new\n0ef4f98825 Remove policy rule for get_network:router:external\n2081910d6d Always create a \"router_extra_attributes\" register per router\n5b64ac9013 Use the neutron-lib constant for PACKET_RATE_LIMIT table number (2)\n9a483f02be Make MechanismDriverContext plugin_context public\n33de608f04 Avoid race condition when deleting trunk bridges\nc5640eee11 Use the neutron-lib constant for PACKET_RATE_LIMIT table number\nbf44e70db6 OVN: Add support for DHCP option \"domain-search\" for IPv4\n25d4e3862b Code move for metadata haproxy\nc3ebefa5f7 Refactor for meter ID Generator\nff66ca9275 Execute functional test \"test_get_all_devices\" alone\n51827d8e78 Remove \"test_live_migration_with_trunk\" execution\na2090e4cbc Doc: Capitalize keystone domain names\nc5ee9f3495 Load the required configuration options in the UT classes\nbb2d706107 [doc] Metadata service supports IPv6 networks\nb7d04d5d92 Code move for metadata signature function\nc6a83988ea Replace Linux Bridge with Open vSwitch in installation docs\n81004125c4 [OVN] Prevent maintenance task fail resource stdattr does not exist\n9e2a0ac058 Use --exclude-regex in tox.ini\n609ebd9504 test: fix docstring validation function\n562e9704f8 Update documentation link for openSUSE index.\nb1714a2b9d Fix some pylint indentation warnings\n55b16d7b7c Fix some pylint indentation warnings\n3afa1b1471 Fix some pylint indentation warnings\n3702575bfe Fix some pylint indentation warnings\nf77c7c9584 Fix some pylint indentation warnings\n291eabb0b6 Fix some pylint indentation warnings\nd1c2d2c4fe dhcp: support multiple segmentations per network\n01b11df1b5 Add tempest job which will run on the oldest Ubuntu release\n0da328fbee Add \"tempest-integrated-networking\" job to the check and gate queues\n1abb77d7a6 Check subnet overlapping after add router interface\n304394ed77 Revert \"Temporary make rally job non voting\"\nc871135cdd Remove xenapi from neutron ml2 config opts\n8fcf00a36d Disable in-band management for bridges before setting up controllers\nc33b47edc7 Do not keep gateway port when notifying for router update\n7c1a894ce5 Nit: network_update in ovs_neutron_agent has a bad LOG\n963f8e30d8 CI: Add tick-tick grenade jobs to check queue\nda335193ef Promote neutron-ovn-rally-task job to be voting in check queue\n8d5985ba6a [Docs] Remove python version from the ci jobs summary\n99339a9704 Clean \"ovn ipv6 only\" jobs templates\nb7b8f7c571 [OVN] Avoid deadlock when cleaning hash ring nodes\ncb332acb29 dhcp: add/use cleanup stale devices API\n1d8e3b79db dhcp: make device manager to clean only not used devices\n4551718bad Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\nee7bfca937 Remove unnecessary bullet points, unify formatting.\n7a5a61616a Remove reference to \u0027all-plugin\u0027 tox environment\n72dd768a12 Revert \"Since OVN 20.06, config is stored in \"Chassis.other_config\"\"\n628442aed7 Spread OVN metadata agent heartbeat response in time\nca2e593177 Change grenade job names suffix to \"skip-level\"\n240f2c6aeb Split Hash Ring probing from the maintenance task\n1920f2a4e3 Add new oslo master CI jobs to the periodic queue\n7b8b797dea Load the required configuration options in the UT classes\n4858315286 Fix misplaced comparison constant warnings\n3d0f1ef4fa Remove some unnecessary list comprehensions\n70a2029ca1 [OVN] Set the default OVN metadata worker number to 0\n34b81ae9b7 Remove OVN gap minimum bandwidth support\n229d798a45 [OVN] Allow only one physical network per bridge\nf43891bf86 [OVN] Allow to execute ``MetadataProxyHandler`` in a local thread\n50eee19723 ovn: Use ovsdb-client to create neutron_pg_drop\n536498a29a Since OVN 20.06, config is stored in \"Chassis.other_config\"\n0c9e7d4e18 Update grenade skip level jobs for new release\nde18dd2c2e [ovn migration] Use ecsda ssh key instead of rsa\nd7b44f7218 Allow to pass EUI64 IP address as fixed ip for the port\na3e68e8f76 Fix ipam_pluggable_backend unit tests module\na8a43c8045 Execute \"IpMonitorTestCase\" tests always inside a namespace\nf8fa909444 Since OVN 20.06, config is stored in \"Chassis.other_config\"\n7b15ad9433 Bump minimum OVN version to v21.06.0\ne4cc40f114 fix: Fix url of Floodlight\n21491efd9f Port provisioning should retry only for VM ports\n178ee6fd3d [Trunk] Update the trunk status with the parent status\n0b2844b8a8 Add .pylintrc to irrelevant-files for non pep8 jobs\n72493d1472 Add missing comma in .pylintrc\n3f20cabc4f Revert \"Move the \"ovn_hash_ring\" clean up to maintenance worker\"\n56f89131dd docs: Remove errant indentation, mark up literals\n2f0919f7df Accept a port deletion with missing port binding information\n1fe4edab80 Open Antelope DB branch\n41a7479d0f Handle several dhcp agents for metadata over ipv6\n819a1bb3e6 Move the \"ovn_hash_ring\" clean up to maintenance worker\ne9b5a27001 Use \"OVNMechDriver\" instance in \"TestOvn[Nb|Sb]IdlNotifyHandler\"\n9eddb132b2 Fix openstack-tox-py39-with-oslo-master job\na765b7acbb [sqlalchemy-20] Add py39 oslo master to experimental\nea11928ba0 Neutron needs oslo.config \u003e\u003d 9\n9a1b5d1e51 Imported Translations from Zanata\n46bc1e7f13 Switch to 2023.1 Python3 unit tests and generic template name\n4914d7fab2 Update master for stable/zed\n639257c77b Make path for tcpdump release-specific\ncca48be73d Imported Translations from Zanata\nf7578e51f2 Collect ovn/ovs logs in singlenode jobs\n9655466763 [OVN] Rate limit the \"Disallow caching\" log from hash ring\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Iefe51ff7ca7e451d4149ad8886896e990d7066e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66b571ee5283e406ccb31b7c48235fbdda20b041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66b571ee5283e406ccb31b7c48235fbdda20b041"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
