)]}'
{"id":"openstack%2Freleases~875450","triplet_id":"openstack%2Freleases~master~I9c7a411aef9b483e5b89cfd3c2c61219cbcf832d","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"6469":{"account":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"last_update":"2023-03-03 11:06:00.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-03 10:34:15.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-03 11:06:00.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-03 11:06:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9c7a411aef9b483e5b89cfd3c2c61219cbcf832d","subject":"Release octavia RC1 for Antelope","status":"MERGED","created":"2023-02-27 12:08:06.000000000","updated":"2023-03-03 11:06:00.000000000","submitted":"2023-03-03 11:06:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875450-antelope-rc1-deadline","meta_rev_id":"12cad2bd9b5346914b14657ea44cec3c100f4d02","_number":875450,"virtual_id_number":875450,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-03 11:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-03-03 09:37:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-03-03 10:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-02 15:24:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-03-03 10:34:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-03-02 13:30:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 12:09:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 12:21:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2023-02-27 12:21:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-02-27 12:21:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-02-27 12:39:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-02 10:49:27.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2023-03-03 09:37:00.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"99542761ac0ae6136defcf220215004369d0a00d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 12:08:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef616bd2794fb137762cc01ae4bd703112ecf587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:39:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2e393d4ae0640ce909e0280fa1bd1ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d388f7c078144b19986f974d6516f95f : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c38b1002340941829e5bb5f6ca0aa3ad : SUCCESS in 5m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6eb9e481318f40c6832053023ddcb701 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1d47a3c98db33b7049958c38c0792790d02fba89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:14:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"95277652261ba11d137dd3d728dcab015a375daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:23:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d77d7fc754044c7891b5f01395f50fce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/744b124a6f1748fe9453a03a0e68c0d4 : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8836e24d83734adeb7fcca332ceecbfb : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e4409b0e01d54f6f9447033e680a2e45 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"b2a04b813fac740d557fad8218788c7521bbab68","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-01 14:31:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1604cde132fc8349c33addaaf2aab858cfd9bb97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 13:28:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1ff2f2171148692b0b971e84dbbc4475e2ed252f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 13:28:19.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9418f7c98ac36c991cabf28a500c6665a95a341e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 13:30:31.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"49f96b3c7e520df17603d7385b704cef2d1526ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 13:43:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51ad4608b8e24042bbf775ddedb7ae66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d06930085ccf478daeabf44e67333ccf : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/951dba4e1feb43649be85e28f9a8ed41 : SUCCESS in 5m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0e0e1cc88d1543029a582fe46032f012 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"c15a44af98f14b6a0f9ca98232cf2ba88765b19d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 15:24:13.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"441ab16d52f0f0b521838c31241ecfb0fc85aef5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-03 09:37:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f02c5dca516fa1e4270de73fcf8b61db2c4b7f75","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-03 10:34:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4f10886ce1aa5516c285061c9b68b1ec496e5e3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 10:34:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4a5dcbb7f5b4a062d96b164d46810410d6c5f321","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 11:05:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27cbc43903d8434d83f4aad3bec670f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4950fa724f8428b9a6e10cbdd6f6f8b : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c490c1bbb03645409dd6c6e908ace147 : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c37ab19461fb44a8a532027aed32c3cb : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"12cad2bd9b5346914b14657ea44cec3c100f4d02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 11:06:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"44ce5f4d832ccc07661024d73ab00b922cfdd948","revisions":{"3d28f1743ba75658f8d2383d024b0d3dca9dafdd":{"kind":"REWORK","_number":1,"created":"2023-02-27 12:08:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/50/875450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/875450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/875450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/875450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/875450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/875450/1"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:08:02.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:08:02.000000000","tz":60},"subject":"Release octavia RC1 for Antelope","message":"Release octavia RC1 for Antelope\n\nThis proposed a RC1 release for octavia to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 11.0.0..b4b6e07f\nc2c59f4c Fix unit tests for python-barbicanclient 5.5.0\n54e1f6f1 Fix job book deletion issue in taskflow\na36e3dda Fix grenade job\nc55839ca Fix wrong assertion methods\nfe65e1ef Imported Translations from Zanata\nff8c3aa7 Fix new pep8/pylint warnings\n7a2df883 Remove support for split listener configuration\n6c731fa2 Filter out details from taskflow logs with v2+jobboard\n33a3d474 fix haproxy_count to display the number of running processes\nb28be698 Fixing tests with tox 4.2.6\nce7c796e Remove unused create_load_balancer_tree code\n41ee7e7a Fix broken tox.ini with tox4\nd0a0f53d Fix listener creation with fully-populated API\n9c8b00bc Add release note about fixing castellan\n1d84f297 Fix typo in docs section title\n454cff58 Optimize HAProxy SSL cache size\nbaefc3b4 Set nohz_full in cpu-pinning element\n0b1b6c58 Amphora vertical scaling optimization in Octavia\nc1ebe18e Add cpu-pinning element\nff1356a6 Ensure options are dumped in debug logs\ne0d21766 Add a newline when writing the server state file\n37ddc6fb Update diskimage-create.sh to set jammy as default Ubuntu version\n06f67f44 Filter ComputeWaitTimeoutException when jobboard is disabled\n344d6d76 Fix prometheus-proxy service name in Red Hat-based distros\ndd5d3cca Fix persistence_granularity default value\n62e1d871 Rename Context to RequestContext\nf5ac714a Pass config to castellan\ne150e032 Add pre-commit\n216cce39 Fix full graph loadbalancer creation if jobboard is disabled\nfa0086cd devstack: use iniset_rpc_backend\n962b4b29 Imported Translations from Zanata\ne10cc8ce Switch to 2023.1 Python3 unit tests and generic template name\n74b209f0 Update master for stable/zed\n69ecdb19 Handle feature compatibility of HAProxy server-state-file option\nabf7b9ef Add conf file for oslo-config-generator\n09015f85 Fix missing help texts for some config options\nce7f27e3 Fix barbican client with application credentials/trusts\n54a778e8 Add *.orig to .gitignore\n55e6001e Fix typo in lvs-masquerade.sh\ncbf79c56 Fix PING health-monitor with recent Centos releases\n4333881c Remove i386 amphora build support\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I9c7a411aef9b483e5b89cfd3c2c61219cbcf832d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3d28f1743ba75658f8d2383d024b0d3dca9dafdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3d28f1743ba75658f8d2383d024b0d3dca9dafdd"}]},"branch":"refs/heads/master"},"dd98bb4bb1553b225679e16ed0be7bc42103bff4":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:14:58.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/50/875450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/875450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/875450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/875450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/875450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/875450/2"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:08:02.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 14:14:58.000000000","tz":0},"subject":"Release octavia RC1 for Antelope","message":"Release octavia RC1 for Antelope\n\nThis proposed a RC1 release for octavia to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 11.0.0..b4b6e07f\nc2c59f4c Fix unit tests for python-barbicanclient 5.5.0\n54e1f6f1 Fix job book deletion issue in taskflow\na36e3dda Fix grenade job\nc55839ca Fix wrong assertion methods\nfe65e1ef Imported Translations from Zanata\nff8c3aa7 Fix new pep8/pylint warnings\n7a2df883 Remove support for split listener configuration\n6c731fa2 Filter out details from taskflow logs with v2+jobboard\n33a3d474 fix haproxy_count to display the number of running processes\nb28be698 Fixing tests with tox 4.2.6\nce7c796e Remove unused create_load_balancer_tree code\n41ee7e7a Fix broken tox.ini with tox4\nd0a0f53d Fix listener creation with fully-populated API\n9c8b00bc Add release note about fixing castellan\n1d84f297 Fix typo in docs section title\n454cff58 Optimize HAProxy SSL cache size\nbaefc3b4 Set nohz_full in cpu-pinning element\n0b1b6c58 Amphora vertical scaling optimization in Octavia\nc1ebe18e Add cpu-pinning element\nff1356a6 Ensure options are dumped in debug logs\ne0d21766 Add a newline when writing the server state file\n37ddc6fb Update diskimage-create.sh to set jammy as default Ubuntu version\n06f67f44 Filter ComputeWaitTimeoutException when jobboard is disabled\n344d6d76 Fix prometheus-proxy service name in Red Hat-based distros\ndd5d3cca Fix persistence_granularity default value\n62e1d871 Rename Context to RequestContext\nf5ac714a Pass config to castellan\ne150e032 Add pre-commit\n216cce39 Fix full graph loadbalancer creation if jobboard is disabled\nfa0086cd devstack: use iniset_rpc_backend\n962b4b29 Imported Translations from Zanata\ne10cc8ce Switch to 2023.1 Python3 unit tests and generic template name\n74b209f0 Update master for stable/zed\n69ecdb19 Handle feature compatibility of HAProxy server-state-file option\nabf7b9ef Add conf file for oslo-config-generator\n09015f85 Fix missing help texts for some config options\nce7f27e3 Fix barbican client with application credentials/trusts\n54a778e8 Add *.orig to .gitignore\n55e6001e Fix typo in lvs-masquerade.sh\ncbf79c56 Fix PING health-monitor with recent Centos releases\n4333881c Remove i386 amphora build support\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I9c7a411aef9b483e5b89cfd3c2c61219cbcf832d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd98bb4bb1553b225679e16ed0be7bc42103bff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd98bb4bb1553b225679e16ed0be7bc42103bff4"}]},"branch":"refs/heads/master"},"44ce5f4d832ccc07661024d73ab00b922cfdd948":{"kind":"REWORK","_number":3,"created":"2023-03-02 13:28:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/50/875450/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/875450/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/875450/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/875450/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/875450/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/875450/3"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:08:02.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 13:26:33.000000000","tz":60},"subject":"Release octavia RC1 for Antelope","message":"Release octavia RC1 for Antelope\n\nThis proposed a RC1 release for octavia to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 11.0.0..8f4cde36\n8f2b48a2 Deprecate user_data_config_drive option\n710e9105 Fix Octavia policies\n8409c06c Fix SQLAlchemy warning about conflict relationship with Tags\n8db655ae Fix text/plain mime type with healthcheck endpoint\nc0e55024 Fix Octavia API HTTP Accept header handling\nc2c59f4c Fix unit tests for python-barbicanclient 5.5.0\n244f44ae Use new get_rpc_client API from oslo.messaging\n54e1f6f1 Fix job book deletion issue in taskflow\na36e3dda Fix grenade job\nc55839ca Fix wrong assertion methods\nfe65e1ef Imported Translations from Zanata\nff8c3aa7 Fix new pep8/pylint warnings\n7a2df883 Remove support for split listener configuration\n6c731fa2 Filter out details from taskflow logs with v2+jobboard\n33a3d474 fix haproxy_count to display the number of running processes\nb28be698 Fixing tests with tox 4.2.6\nce7c796e Remove unused create_load_balancer_tree code\n41ee7e7a Fix broken tox.ini with tox4\nd0a0f53d Fix listener creation with fully-populated API\n9c8b00bc Add release note about fixing castellan\n1d84f297 Fix typo in docs section title\n454cff58 Optimize HAProxy SSL cache size\nbaefc3b4 Set nohz_full in cpu-pinning element\n0b1b6c58 Amphora vertical scaling optimization in Octavia\nc1ebe18e Add cpu-pinning element\nff1356a6 Ensure options are dumped in debug logs\ne0d21766 Add a newline when writing the server state file\n37ddc6fb Update diskimage-create.sh to set jammy as default Ubuntu version\n06f67f44 Filter ComputeWaitTimeoutException when jobboard is disabled\n344d6d76 Fix prometheus-proxy service name in Red Hat-based distros\ndd5d3cca Fix persistence_granularity default value\n62e1d871 Rename Context to RequestContext\nf5ac714a Pass config to castellan\ne150e032 Add pre-commit\n216cce39 Fix full graph loadbalancer creation if jobboard is disabled\nfa0086cd devstack: use iniset_rpc_backend\nb22e0726 Fix image creation for stable branches\n962b4b29 Imported Translations from Zanata\ne10cc8ce Switch to 2023.1 Python3 unit tests and generic template name\n74b209f0 Update master for stable/zed\n69ecdb19 Handle feature compatibility of HAProxy server-state-file option\nabf7b9ef Add conf file for oslo-config-generator\n09015f85 Fix missing help texts for some config options\nce7f27e3 Fix barbican client with application credentials/trusts\n54a778e8 Add *.orig to .gitignore\n55e6001e Fix typo in lvs-masquerade.sh\ncbf79c56 Fix PING health-monitor with recent Centos releases\n4333881c Remove i386 amphora build support\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I9c7a411aef9b483e5b89cfd3c2c61219cbcf832d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/44ce5f4d832ccc07661024d73ab00b922cfdd948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/44ce5f4d832ccc07661024d73ab00b922cfdd948"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
