)]}'
{"id":"openstack%2Freleases~875466","triplet_id":"openstack%2Freleases~master~I4d7bc04da93735c006c09c8b349feba9d5ea614f","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-03-01 10:48:08.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-01 10:34:27.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2023-03-01 06:04:54.000000000","reason":"\u003cGERRIT_ACCOUNT_6282\u003e replied on the change","reason_account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}},"hashtags":[],"change_id":"I4d7bc04da93735c006c09c8b349feba9d5ea614f","subject":"Release skyline-console RC1 for Antelope","status":"MERGED","created":"2023-02-27 12:20:32.000000000","updated":"2023-03-01 10:48:08.000000000","submitted":"2023-03-01 10:48:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875466-antelope-rc1-deadline","meta_rev_id":"67a09c351668d30f026480d8036a09db7862da6e","_number":875466,"virtual_id_number":875466,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-01 10:48:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-03-01 04:47:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-03-01 08:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-03-01 06:04:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":2,"date":"2023-03-01 10:34:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":1,"date":"2023-03-01 10:34:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-03-01 02:39:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 12:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 12:40:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 13:21:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2023-02-27 13:21:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2023-02-28 12:56:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-02-28 13:22:20.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-03-01 08:49:10.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1b70f6710a9b50d8c3585f87169d0187aaab8096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 12:20:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5647f4f4795bd424d28817bc616b73469787356b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:40:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfe0ca30b7d448f3a329e0bc05ca01b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da8ddefa4844412a2c0be1854a8695a : SUCCESS in 8m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f18569ea8e1424aad2c8d7f3881249a : SUCCESS in 5m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/68582bd8236446cba581bac9ce1a1c34 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"00214a53fe458085b52e45a045adb3692ca284e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:02:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d86220c09df25cd09185e187675dbcb384f2b298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:12:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9e590aa84064bb2a230bcded44e2d2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16e957e553cc40fe902e16db11226494 : SUCCESS in 7m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/031ef981b1d14fb6a7aa78aeed7b7632 : SUCCESS in 5m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8d16267c60c142318f6bc3892f135e06 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e5c3cf186b382551a13bca74c662223aab2237ac","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-02-27 14:43:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b7a06ac1164adde4d7940f44076be7d345af3fd8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:44:11.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"21f9044a641900201bf9770ab45f1a8daf79a9c0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-02-28 12:56:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a4e262bb097612855308280d564f6db91da70497","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-28 13:22:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9b04301a08a8f557ebf7ff5d9fc1f8e612cb7622","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-03-01 02:38:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"de82f1960a84efc0c768a214289cfdea973ad1a0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 02:39:31.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"372a1b1e874d5ff2895bd0c42778b9e2decf640e","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-03-01 02:39:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"017e0e85c4c098791742da24b1a59f68fd95fa4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 02:53:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/875e63e42b2847eeb846e69f6d7634bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a85ac7b31e8f49e9bb6d39e6ae37603c : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c68a66f5e14b449698214f287d2f5a27 : SUCCESS in 5m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/681ee5fdd4c14abe946c19cd98f8ba39 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2c274f67970c99e406fe6a15d80515fffa144033","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-03-01 04:47:12.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c7f63813c3ccba97bfb5dfa7e003ba54568d69cc","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2023-03-01 06:04:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7fd83bc34751c29a983a282ced6cd1762899391b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-01 08:49:10.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"56ef94191d30d25f90032f3a0b34b6695f733852","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-01 10:34:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f7b6f6c1814f219fbca995c29445ae3f6e5b577a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 10:34:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"71d4d828e15e867089d8f212c2500977719f7e51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 10:48:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c412378dcd1540f5b2ec4b9eb2953f40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d468bc740afe410589cb45a53dba0d4f : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e43b02db1259497d94d1b3e9882e7788 : SUCCESS in 7m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e75f1e91957413b8a3455665a29b8c9 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":3},{"id":"67a09c351668d30f026480d8036a09db7862da6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 10:48:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8501bdf6834c03c6ec5f3fa8bf6307be78f9bc91","revisions":{"864dee5fe39de3b1f25a48ee4c84d75400ddce03":{"kind":"REWORK","_number":1,"created":"2023-02-27 12:20:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/66/875466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/875466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/875466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/875466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/875466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/875466/1"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:20:28.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:20:28.000000000","tz":60},"subject":"Release skyline-console RC1 for Antelope","message":"Release skyline-console RC1 for Antelope\n\nThis proposed a RC1 release for skyline-console to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 1.0.0.0rc1..929842c8\n929842c8 fix: add default value and tip when create zun container\n635f8292 fix: fix the rebuild container and ports of zun\n01ea21d1 fix: fix the zun-ui\n974427c3 fix: fix the resize of magnum cluster\nd5d14d76 feature: Support upgrade cluster and show quota info when resize cluster\n5a45f54d fix: Modify label names to make them consistent\nc5953f17 fix keypair showing in magnum service\n021bde48 fix: fix the error of compute service state\n2ebb2694 feat: add attributes to the ModalAction\n417e729e feat: support textarea in the KeyValueInput component\nd80c23b2 fix: fix the title of zun container\n6c34ed67 fix: fix the magnum template detail info\na347c373 feature: support ip address, ports and networks in list and detail\n317a81ac feature: support search filters and fix image in zun ui\n5f5b8c3c feature: support some new params and fix some issues of zun ui\na11a0258 fix: fix the download data.\n0568a3ea fix deleting and detail info showing of cluster\nefc5fd88 fix: eliminate whitespace in passenv values\n583386d2 fix: seprate clusters and templates by different project\n331c2ddd fix: fix the project column\ne03182b0 fix: fix the magnum quotai and other infos\nd48aca89 fix: reset the subnet list when create cluster instance\n7b6e8253 fix: fix the delete operation of cluster instance\nf4a24d04 fix: fix the getValue of yaml\n2d437e5b feat: support custom render in the TabDetail component\na307f457 fix: fix the magnum service\nf384bd4b feature: Add listener input event for instance-volume component\n124fdc3c fix: add disabledRemoveFunc prop in AddSelect component\n3ada93aa fix the validator of cluster name and labels\n594db176 fix: add disableEditKeys prop in AddSelect component\nc7e19491 fix: support name validator and fix placeholder of magnum service\n77df0547 fix: Support search filter for cluster and template.\n228ccff0 fix: Improve quota in magnum service\nef353362 feat: support component in the table header\n7a4d17ee feature: Support quota in magnum service\n719e26eb fix: add middleComponent prop in KeyValueInput component\nb4f23fd7 fix: fix health_status in magnum cluster\n0c3a946d fix: Move the keypair to node spec when create cluster instance\nac1edd6e Fix keypair in magnum\n8401c5b5 fix: fix the instance-volume in form items\n72b81c64 Fix: fix the image store in magnum node spec\n463ced26 fix: Fix the store path in magnum and required name in zun container\nf7ae6ff9 fix: Fix quick creating keypair in creating instance\n801bf664 fix: fix the router path and logs in zun container\nc4987265 feature: Support quick creating keypair in Magnum service\nc1eb34a5 fix: add confirm resize/migrate button in the administrator\nbd30f8c3 feat: add success and error status in Status component\n47b3cead feature: Support magnum in administrator platform\n9df7b6ca fix: fix the i18n of the status of vpn ipsec connection\n638ac6f2 feature: Add translation for monitor center\n7e30ad69 feat: update the click area of links in the global navigation\nf89ffe61 Fix the state and default value when create cluster instance\n5eab747f Fix the cluster and template of magnum\n6ed2fb1a feat: support search in the global navigation\ne3543ed1 feat: update menu in the layout\n985c03af feat: add global navigation\n5ba32d6d feature: Optimization of Magnum\n5a476a09 feat: add props in BaseForm\n72eb092e Remove unicode literal from code\neaf68bea Remove unicode literal from code\n442d0557 feature: Support custom button for the upload\n0849ecea Fix : Update labels , filters , keypairs at Magnum Service\n87fa14b0 fix: When the storage backends usage is 0 displayed error as \u0027-\u0027\n4639f2a8 fix: The scale of the x-axis is not updated in time\n544f022a chore: Replace https with http as default\n3b1ddc54 fix: Calculate quota based on the data disk of the selected instance snapshot\n33f90ae3 fix: No data disk is displayed on the ConfirmStep\nfdd6dbb5 fix: Not show the data disk when creating vm by Instance snapshot\nb7a8b6f3 feat: Support download file progress and abort download\n27fb2b2d fix: fix the row auto selected when click the copy button edge\n47c34e11 fix: update the qos policy column in the floating ip list page\n73e6153e fix: update the project name display in the rebuild instance action\n4edd0784 fix: fix the download data in the all networks tab\n2e22275d fix: update the id/name column of listener members\na29d9dac fix: update the click area of the id link\na17edb45 fix: fix the row auto selected when copy id\n7c8f9f7e feat: Support for custom prometheus requests\n7b87e7e7 fix: set the default language is Engilsh\n3078eace fix: Fix the unit transform in prometheus monitor\n265e3c79 feat: remove id link in the name column in the role list page\nfb3078f7 feat: update id render to support number format\n27f127f8 fix: fix the category options in the flavor list page\n8398406d docs: add the release note\n11501582 feat: update name column in the keypair list page\nd158a05b feat: display the id/name column in the table in a unified style\nf9cc9dea docs: Change the launchpad url for skyline-console\n6371effd feat: support id/name in non-bold form\nc3abea15 feat: update the generic projectRender\n26d6935b feat: update project id/name column display\n88eb2ad8 feat: update the generic idNameColumn\n3026f4fc feat: update id/name column display\nb927f6c5 Hide the certificate info in administrator platform\n574580a7 fix: modify padding of security groups on instance detail english page\n2f286ce3 fix: add Specify Physical Node label of manually specify when create instance\nffd7b1a4 feat: update e2e zuul tasks\n5fc06358 feat: support flavor to add cpu/memory search filters\nf9cc92a0 Fix edit member of lb listener\n79d41791 fix: add vm state check for modify instance tags\n069d45af Fix the name of Bare Metal Nodes\ncdd68637 fix: security groups style is error on instance detail english page\nbe289f56 fix: fix the pagination in the vpns pages\ned8c6e69 fix: set the type required when create vpn endpoint group\na6a730f7 fix: fix select static router\na92c6aa9 fix associate floating IP in the port list if allowed\nc1fb03c4 Fix Edit Qos Policy\nb0c629ef fix the QoS policy list\nfcb2856c fix: edit image\u0027s min disk\n1cdecff6 fix: remove tags filter in the instance list page\n6482bb0b fix: Fix the disable property of check-box-group\n966a6546 fix: Change the unit to \u0027TiB,GiB....\u0027\nc7c7b8b8 fix: validate error when clear input-name\n19d0cf20 fix: fix octavia request headers\nf20b28c6 fix: The loading cannot be canceled\nfdcf5c14 fix: create instance by given image/instance snapshot\n06af9d6f fix: fix fetch subnets\n849db0f7 fix: set AZ required when create aggregate\nc7556cfd fix: Fix the unit when upload file\n92b91ba0 fix: fix edit image\naef19528 Update master for stable/zed\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I4d7bc04da93735c006c09c8b349feba9d5ea614f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/864dee5fe39de3b1f25a48ee4c84d75400ddce03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/864dee5fe39de3b1f25a48ee4c84d75400ddce03"}]},"branch":"refs/heads/master"},"72d9da56d9c80c833728543e3be139fa292a86f7":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:02:17.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/66/875466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/875466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/875466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/875466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/875466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/875466/2"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:20:28.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 14:02:16.000000000","tz":0},"subject":"Release skyline-console RC1 for Antelope","message":"Release skyline-console RC1 for Antelope\n\nThis proposed a RC1 release for skyline-console to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 1.0.0.0rc1..929842c8\n929842c8 fix: add default value and tip when create zun container\n635f8292 fix: fix the rebuild container and ports of zun\n01ea21d1 fix: fix the zun-ui\n974427c3 fix: fix the resize of magnum cluster\nd5d14d76 feature: Support upgrade cluster and show quota info when resize cluster\n5a45f54d fix: Modify label names to make them consistent\nc5953f17 fix keypair showing in magnum service\n021bde48 fix: fix the error of compute service state\n2ebb2694 feat: add attributes to the ModalAction\n417e729e feat: support textarea in the KeyValueInput component\nd80c23b2 fix: fix the title of zun container\n6c34ed67 fix: fix the magnum template detail info\na347c373 feature: support ip address, ports and networks in list and detail\n317a81ac feature: support search filters and fix image in zun ui\n5f5b8c3c feature: support some new params and fix some issues of zun ui\na11a0258 fix: fix the download data.\n0568a3ea fix deleting and detail info showing of cluster\nefc5fd88 fix: eliminate whitespace in passenv values\n583386d2 fix: seprate clusters and templates by different project\n331c2ddd fix: fix the project column\ne03182b0 fix: fix the magnum quotai and other infos\nd48aca89 fix: reset the subnet list when create cluster instance\n7b6e8253 fix: fix the delete operation of cluster instance\nf4a24d04 fix: fix the getValue of yaml\n2d437e5b feat: support custom render in the TabDetail component\na307f457 fix: fix the magnum service\nf384bd4b feature: Add listener input event for instance-volume component\n124fdc3c fix: add disabledRemoveFunc prop in AddSelect component\n3ada93aa fix the validator of cluster name and labels\n594db176 fix: add disableEditKeys prop in AddSelect component\nc7e19491 fix: support name validator and fix placeholder of magnum service\n77df0547 fix: Support search filter for cluster and template.\n228ccff0 fix: Improve quota in magnum service\nef353362 feat: support component in the table header\n7a4d17ee feature: Support quota in magnum service\n719e26eb fix: add middleComponent prop in KeyValueInput component\nb4f23fd7 fix: fix health_status in magnum cluster\n0c3a946d fix: Move the keypair to node spec when create cluster instance\nac1edd6e Fix keypair in magnum\n8401c5b5 fix: fix the instance-volume in form items\n72b81c64 Fix: fix the image store in magnum node spec\n463ced26 fix: Fix the store path in magnum and required name in zun container\nf7ae6ff9 fix: Fix quick creating keypair in creating instance\n801bf664 fix: fix the router path and logs in zun container\nc4987265 feature: Support quick creating keypair in Magnum service\nc1eb34a5 fix: add confirm resize/migrate button in the administrator\nbd30f8c3 feat: add success and error status in Status component\n47b3cead feature: Support magnum in administrator platform\n9df7b6ca fix: fix the i18n of the status of vpn ipsec connection\n638ac6f2 feature: Add translation for monitor center\n7e30ad69 feat: update the click area of links in the global navigation\nf89ffe61 Fix the state and default value when create cluster instance\n5eab747f Fix the cluster and template of magnum\n6ed2fb1a feat: support search in the global navigation\ne3543ed1 feat: update menu in the layout\n985c03af feat: add global navigation\n5ba32d6d feature: Optimization of Magnum\n5a476a09 feat: add props in BaseForm\n72eb092e Remove unicode literal from code\neaf68bea Remove unicode literal from code\n442d0557 feature: Support custom button for the upload\n0849ecea Fix : Update labels , filters , keypairs at Magnum Service\n87fa14b0 fix: When the storage backends usage is 0 displayed error as \u0027-\u0027\n4639f2a8 fix: The scale of the x-axis is not updated in time\n544f022a chore: Replace https with http as default\n3b1ddc54 fix: Calculate quota based on the data disk of the selected instance snapshot\n33f90ae3 fix: No data disk is displayed on the ConfirmStep\nfdd6dbb5 fix: Not show the data disk when creating vm by Instance snapshot\nb7a8b6f3 feat: Support download file progress and abort download\n27fb2b2d fix: fix the row auto selected when click the copy button edge\n47c34e11 fix: update the qos policy column in the floating ip list page\n73e6153e fix: update the project name display in the rebuild instance action\n4edd0784 fix: fix the download data in the all networks tab\n2e22275d fix: update the id/name column of listener members\na29d9dac fix: update the click area of the id link\na17edb45 fix: fix the row auto selected when copy id\n7c8f9f7e feat: Support for custom prometheus requests\n7b87e7e7 fix: set the default language is Engilsh\n3078eace fix: Fix the unit transform in prometheus monitor\n265e3c79 feat: remove id link in the name column in the role list page\nfb3078f7 feat: update id render to support number format\n27f127f8 fix: fix the category options in the flavor list page\n8398406d docs: add the release note\n11501582 feat: update name column in the keypair list page\nd158a05b feat: display the id/name column in the table in a unified style\nf9cc9dea docs: Change the launchpad url for skyline-console\n6371effd feat: support id/name in non-bold form\nc3abea15 feat: update the generic projectRender\n26d6935b feat: update project id/name column display\n88eb2ad8 feat: update the generic idNameColumn\n3026f4fc feat: update id/name column display\nb927f6c5 Hide the certificate info in administrator platform\n574580a7 fix: modify padding of security groups on instance detail english page\n2f286ce3 fix: add Specify Physical Node label of manually specify when create instance\nffd7b1a4 feat: update e2e zuul tasks\n5fc06358 feat: support flavor to add cpu/memory search filters\nf9cc92a0 Fix edit member of lb listener\n79d41791 fix: add vm state check for modify instance tags\n069d45af Fix the name of Bare Metal Nodes\ncdd68637 fix: security groups style is error on instance detail english page\nbe289f56 fix: fix the pagination in the vpns pages\ned8c6e69 fix: set the type required when create vpn endpoint group\na6a730f7 fix: fix select static router\na92c6aa9 fix associate floating IP in the port list if allowed\nc1fb03c4 Fix Edit Qos Policy\nb0c629ef fix the QoS policy list\nfcb2856c fix: edit image\u0027s min disk\n1cdecff6 fix: remove tags filter in the instance list page\n6482bb0b fix: Fix the disable property of check-box-group\n966a6546 fix: Change the unit to \u0027TiB,GiB....\u0027\nc7c7b8b8 fix: validate error when clear input-name\n19d0cf20 fix: fix octavia request headers\nf20b28c6 fix: The loading cannot be canceled\nfdcf5c14 fix: create instance by given image/instance snapshot\n06af9d6f fix: fix fetch subnets\n849db0f7 fix: set AZ required when create aggregate\nc7556cfd fix: Fix the unit when upload file\n92b91ba0 fix: fix edit image\naef19528 Update master for stable/zed\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I4d7bc04da93735c006c09c8b349feba9d5ea614f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/72d9da56d9c80c833728543e3be139fa292a86f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/72d9da56d9c80c833728543e3be139fa292a86f7"}]},"branch":"refs/heads/master"},"8501bdf6834c03c6ec5f3fa8bf6307be78f9bc91":{"kind":"REWORK","_number":3,"created":"2023-03-01 02:38:18.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/66/875466/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/875466/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/875466/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/875466/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/875466/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/875466/3"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:20:28.000000000","tz":60},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-03-01 02:38:04.000000000","tz":480},"subject":"Release skyline-console RC1 for Antelope","message":"Release skyline-console RC1 for Antelope\n\nThis proposed a RC1 release for skyline-console to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 1.0.0.0rc1..0a192be4\n0a192be4 feature: Support attach network and detach network for zun container\n929842c8 fix: add default value and tip when create zun container\n635f8292 fix: fix the rebuild container and ports of zun\n01ea21d1 fix: fix the zun-ui\n974427c3 fix: fix the resize of magnum cluster\nd5d14d76 feature: Support upgrade cluster and show quota info when resize cluster\n5a45f54d fix: Modify label names to make them consistent\nc5953f17 fix keypair showing in magnum service\n021bde48 fix: fix the error of compute service state\n2ebb2694 feat: add attributes to the ModalAction\n417e729e feat: support textarea in the KeyValueInput component\nd80c23b2 fix: fix the title of zun container\n6c34ed67 fix: fix the magnum template detail info\na347c373 feature: support ip address, ports and networks in list and detail\n317a81ac feature: support search filters and fix image in zun ui\n5f5b8c3c feature: support some new params and fix some issues of zun ui\na11a0258 fix: fix the download data.\n0568a3ea fix deleting and detail info showing of cluster\nefc5fd88 fix: eliminate whitespace in passenv values\n583386d2 fix: seprate clusters and templates by different project\n331c2ddd fix: fix the project column\ne03182b0 fix: fix the magnum quotai and other infos\nd48aca89 fix: reset the subnet list when create cluster instance\n7b6e8253 fix: fix the delete operation of cluster instance\nf4a24d04 fix: fix the getValue of yaml\n2d437e5b feat: support custom render in the TabDetail component\na307f457 fix: fix the magnum service\nf384bd4b feature: Add listener input event for instance-volume component\n124fdc3c fix: add disabledRemoveFunc prop in AddSelect component\n3ada93aa fix the validator of cluster name and labels\n594db176 fix: add disableEditKeys prop in AddSelect component\nc7e19491 fix: support name validator and fix placeholder of magnum service\n77df0547 fix: Support search filter for cluster and template.\n228ccff0 fix: Improve quota in magnum service\nef353362 feat: support component in the table header\n7a4d17ee feature: Support quota in magnum service\n719e26eb fix: add middleComponent prop in KeyValueInput component\nb4f23fd7 fix: fix health_status in magnum cluster\n0c3a946d fix: Move the keypair to node spec when create cluster instance\nac1edd6e Fix keypair in magnum\n8401c5b5 fix: fix the instance-volume in form items\n72b81c64 Fix: fix the image store in magnum node spec\n463ced26 fix: Fix the store path in magnum and required name in zun container\nf7ae6ff9 fix: Fix quick creating keypair in creating instance\n801bf664 fix: fix the router path and logs in zun container\nc4987265 feature: Support quick creating keypair in Magnum service\nc1eb34a5 fix: add confirm resize/migrate button in the administrator\nbd30f8c3 feat: add success and error status in Status component\n47b3cead feature: Support magnum in administrator platform\n9df7b6ca fix: fix the i18n of the status of vpn ipsec connection\n638ac6f2 feature: Add translation for monitor center\n7e30ad69 feat: update the click area of links in the global navigation\nf89ffe61 Fix the state and default value when create cluster instance\n5eab747f Fix the cluster and template of magnum\n6ed2fb1a feat: support search in the global navigation\ne3543ed1 feat: update menu in the layout\n985c03af feat: add global navigation\n5ba32d6d feature: Optimization of Magnum\n5a476a09 feat: add props in BaseForm\n72eb092e Remove unicode literal from code\neaf68bea Remove unicode literal from code\n442d0557 feature: Support custom button for the upload\n0849ecea Fix : Update labels , filters , keypairs at Magnum Service\n87fa14b0 fix: When the storage backends usage is 0 displayed error as \u0027-\u0027\n4639f2a8 fix: The scale of the x-axis is not updated in time\n544f022a chore: Replace https with http as default\n3b1ddc54 fix: Calculate quota based on the data disk of the selected instance snapshot\n33f90ae3 fix: No data disk is displayed on the ConfirmStep\nfdd6dbb5 fix: Not show the data disk when creating vm by Instance snapshot\nb7a8b6f3 feat: Support download file progress and abort download\n27fb2b2d fix: fix the row auto selected when click the copy button edge\n47c34e11 fix: update the qos policy column in the floating ip list page\n73e6153e fix: update the project name display in the rebuild instance action\n4edd0784 fix: fix the download data in the all networks tab\n2e22275d fix: update the id/name column of listener members\na29d9dac fix: update the click area of the id link\na17edb45 fix: fix the row auto selected when copy id\n7c8f9f7e feat: Support for custom prometheus requests\n7b87e7e7 fix: set the default language is Engilsh\n3078eace fix: Fix the unit transform in prometheus monitor\n265e3c79 feat: remove id link in the name column in the role list page\nfb3078f7 feat: update id render to support number format\n27f127f8 fix: fix the category options in the flavor list page\n8398406d docs: add the release note\n11501582 feat: update name column in the keypair list page\nd158a05b feat: display the id/name column in the table in a unified style\nf9cc9dea docs: Change the launchpad url for skyline-console\n6371effd feat: support id/name in non-bold form\nc3abea15 feat: update the generic projectRender\n26d6935b feat: update project id/name column display\n88eb2ad8 feat: update the generic idNameColumn\n3026f4fc feat: update id/name column display\nb927f6c5 Hide the certificate info in administrator platform\n574580a7 fix: modify padding of security groups on instance detail english page\n2f286ce3 fix: add Specify Physical Node label of manually specify when create instance\nffd7b1a4 feat: update e2e zuul tasks\n5fc06358 feat: support flavor to add cpu/memory search filters\nf9cc92a0 Fix edit member of lb listener\n79d41791 fix: add vm state check for modify instance tags\n069d45af Fix the name of Bare Metal Nodes\ncdd68637 fix: security groups style is error on instance detail english page\nbe289f56 fix: fix the pagination in the vpns pages\ned8c6e69 fix: set the type required when create vpn endpoint group\na6a730f7 fix: fix select static router\na92c6aa9 fix associate floating IP in the port list if allowed\nc1fb03c4 Fix Edit Qos Policy\nb0c629ef fix the QoS policy list\nfcb2856c fix: edit image\u0027s min disk\n1cdecff6 fix: remove tags filter in the instance list page\n6482bb0b fix: Fix the disable property of check-box-group\n966a6546 fix: Change the unit to \u0027TiB,GiB....\u0027\nc7c7b8b8 fix: validate error when clear input-name\n19d0cf20 fix: fix octavia request headers\nf20b28c6 fix: The loading cannot be canceled\nfdcf5c14 fix: create instance by given image/instance snapshot\n06af9d6f fix: fix fetch subnets\n849db0f7 fix: set AZ required when create aggregate\nc7556cfd fix: Fix the unit when upload file\n92b91ba0 fix: fix edit image\naef19528 Update master for stable/zed\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I4d7bc04da93735c006c09c8b349feba9d5ea614f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8501bdf6834c03c6ec5f3fa8bf6307be78f9bc91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8501bdf6834c03c6ec5f3fa8bf6307be78f9bc91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
