)]}'
{"id":"openstack%2Freleases~875467","triplet_id":"openstack%2Freleases~master~I89f7f650341a16a1aa9980ab4c02788b10532ea1","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-03-07 08:56:13.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-07 09:14:39.000000000","reason":"Change was submitted"},"14107":{"account":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"last_update":"2023-03-03 02:39:47.000000000","reason":"\u003cGERRIT_ACCOUNT_14107\u003e replied on the change","reason_account":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}},"hashtags":[],"change_id":"I89f7f650341a16a1aa9980ab4c02788b10532ea1","subject":"Release solum-dashboard RC1 for Antelope","status":"MERGED","created":"2023-02-27 12:21:30.000000000","updated":"2023-03-07 09:14:39.000000000","submitted":"2023-03-07 09:14:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875467-antelope-rc1-deadline","meta_rev_id":"bbce506b2c41d7faf9dc23de2197e19d999bb02d","_number":875467,"virtual_id_number":875467,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 09:14:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"date":"2023-03-06 08:42:22.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-03-07 08:56:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-03-03 10:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2023-03-07 03:33:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"date":"2023-03-06 08:42:22.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-03-07 08:56:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-03-07 03:34:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 12:22:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 12:33:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 13:22:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2023-03-03 09:38:40.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-03-07 08:56:13.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"27d435e30c681d953b4b8f516d7b462cbc3873b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 12:21:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d62f5bc973bcc5a3f0365513a1e54a60796499c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:33:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25ed9e8bf94e46bf95ebe1016c8c37cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b11ce5ef9764bcd9fa666a9c8043dbf : SUCCESS in 6m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e22bec7ba8a4d8eb72a5bb9ceec21a6 : SUCCESS in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7c3fae1e4352459fa5637745f5f1594f : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4f48e448483dc1bad35e537b7d7ef20c49f53c0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:00:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5ca445bec03fb197690fb56850419908193074d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:25:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df245a3e38fd441e8f637ed9e5e6acc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5df40ce127d447aaa48a72578b6f75e : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f0e79b59f054f23b2417a70308d13dd : SUCCESS in 7m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5c125bbd6cd0491aafd00712b33799d5 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"77411b2852a890f0d7bc37c73ad8459dfd68b2b1","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2023-03-03 02:39:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e7990c205a2426af0314814ee9e7e87d8ccc0f2e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 02:41:14.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"45989a32e8a6b0de65fb9b2752d889ff5e92c3d0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-03 09:38:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3c51a9082c36bb83bd7431407bd6f21e3193a283","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-03 09:52:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3b8a735e4fdaf4bf9fa065cacfd5f9aed582c4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 10:14:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1055962af9c841debc1d1065ce96d21b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c3e146ae134f478fa0370283ea77ce : SUCCESS in 7m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfe435c786eb486ea8c0db6239600bfe : SUCCESS in 4m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/279e95e1de4a4eb4afe45e95faa1bfd6 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":3},{"id":"28ecd6024978334192c0f7ffe5f37dd82c4d6f58","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-03 10:31:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"eb30bd54d2563c37afaf4a9d2aa1ea42d3a01250","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-06 08:34:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6111d9e173ce9eeebecf03d6aa4bed74ef0d0a11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 08:34:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f2c12e5617f095a285f5d14fa8c5547664a5396c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-06 08:35:33.000000000","message":"Patch Set 3: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fea9d47fb32bdc21a2e0c9250912d278e3dec078","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 08:35:45.000000000","message":"Patch Set 3: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":3},{"id":"c7cb71179d5a7ca8e3cb62b887533d903ecc0c45","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-06 08:42:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fe3d9720c3923f9925a02eba68edd0fff0a1e3d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 08:50:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2339e9ef0e624bd4a16135626b4454b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d66cf1408d4c1290209233afd4ac6e : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e90d52bf627f42ea90fe20af820ea805 : SUCCESS in 4m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e8e6a4758df446978c1ff95de850e6c2 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"6d8e6f4d557f11e9570c1a42750cd1d1706fda3f","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2023-03-07 03:33:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9319a3e86a23dc66cf46f8c3de080bbb7aa8947a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 03:34:03.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"134e9d78c868709eb0af32acfdadbd80993d9536","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-07 08:56:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5a60b0b4cc767051c332986f4f05fbe774db64f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 08:56:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ab7935e93ad8129e7e7a29e6caec18f3dea7b875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 09:14:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0ddebef2dde4160a5d784fcaaa068fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/474817e461d74061a89b2ee6c553522d : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657f35b546594a56b7bf2a1e6dc9af99 : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8a496d773c3465994f8b75a00690d17 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":3},{"id":"bbce506b2c41d7faf9dc23de2197e19d999bb02d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 09:14:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"26a454db36b9c83f826a6600ba25ab08970a3495","revisions":{"58cf18ce9e97f857d10a2695d554f5ad78310edc":{"kind":"REWORK","_number":1,"created":"2023-02-27 12:21:30.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/67/875467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/875467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/875467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/875467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/875467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/875467/1"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:21:26.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:21:26.000000000","tz":60},"subject":"Release solum-dashboard RC1 for Antelope","message":"Release solum-dashboard RC1 for Antelope\n\nThis proposed a RC1 release for solum-dashboard to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 9.0.0..e330c0b\ne330c0b Switch to 2023.1 Python3 unit tests and generic template name\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I89f7f650341a16a1aa9980ab4c02788b10532ea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/58cf18ce9e97f857d10a2695d554f5ad78310edc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/58cf18ce9e97f857d10a2695d554f5ad78310edc"}]},"branch":"refs/heads/master"},"105a3700ecb6433f8b268f2740ed34bbe38e966e":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:00:30.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/67/875467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/875467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/875467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/875467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/875467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/875467/2"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:21:26.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 14:00:29.000000000","tz":0},"subject":"Release solum-dashboard RC1 for Antelope","message":"Release solum-dashboard RC1 for Antelope\n\nThis proposed a RC1 release for solum-dashboard to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 9.0.0..e330c0b\ne330c0b Switch to 2023.1 Python3 unit tests and generic template name\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I89f7f650341a16a1aa9980ab4c02788b10532ea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/105a3700ecb6433f8b268f2740ed34bbe38e966e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/105a3700ecb6433f8b268f2740ed34bbe38e966e"}]},"branch":"refs/heads/master"},"26a454db36b9c83f826a6600ba25ab08970a3495":{"kind":"REWORK","_number":3,"created":"2023-03-03 09:52:21.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/875467/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/875467/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/875467/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/875467/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/875467/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/875467/3"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:21:26.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-03-03 09:51:45.000000000","tz":60},"subject":"Release solum-dashboard RC1 for Antelope","message":"Release solum-dashboard RC1 for Antelope\n\nThis proposed a RC1 release for solum-dashboard to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I89f7f650341a16a1aa9980ab4c02788b10532ea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/26a454db36b9c83f826a6600ba25ab08970a3495"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/26a454db36b9c83f826a6600ba25ab08970a3495"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
