)]}'
{"id":"openstack%2Freleases~875471","triplet_id":"openstack%2Freleases~master~Ic1f77c651d1f7c315f86d213561878dba1e58e57","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2023-03-06 09:03:50.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-06 09:03:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic1f77c651d1f7c315f86d213561878dba1e58e57","subject":"Release tacker RC1 for Antelope","status":"MERGED","created":"2023-02-27 12:24:01.000000000","updated":"2023-03-06 09:03:50.000000000","submitted":"2023-03-06 09:03:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875471-antelope-rc1-deadline","meta_rev_id":"be6a72b4276fee210faca211fc59cbfa94d537de","_number":875471,"virtual_id_number":875471,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-06 09:03:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-03-06 08:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-03-03 15:04:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2023-03-03 15:01:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-03-06 08:37:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-03-03 14:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 12:25:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 12:33:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 13:22:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2023-03-03 15:01:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-03-06 08:37:51.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"acc02753e570b7cac874d94b784bbe13899a5381","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 12:24:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67560290ffa18ceb90891c5dd90991934f353ab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:33:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/202429f90a794a8ea3b92964a37facf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3ef5ee7bfdc44428464f02d31b3ff0d : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c17450dbf97848c3aeb45fe8cf9bc070 : SUCCESS in 5m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/af5cab46ea6d4bb097912c812e9dabc4 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0b997b951a0b8830dbd479653d5e86ddd4acf606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:00:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fdd041938586e087e30d7d5a9c68afbdcaf3e13c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:11:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b3a55faa874983aa49bcac85e71113\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43582ee8ed6649c4b6e29a22c1fbcac0 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93228c8f5a614ee8a95cc989c704c5bc : SUCCESS in 5m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/079bbec024994d65894a5bb9d3df3ff7 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3198f9ea1df7dcb61fb72f50e1adf9eedbeefebf","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-03-03 03:06:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0fdaec61dc6aff4b7883adbf271ebbc6c9194ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-03-03 14:42:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c2690bcbc5d9d1464b5dc3fc3aa0d98b6b5772f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 14:43:33.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"c8b3b51eee99a85781044ef94e7fea0a5eafe963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 14:56:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53bf7c153f254a0ba6b06110477ae44e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bcc405cf8ce4d7db7b0d4261892ce83 : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3c93690cd8740ea92645441cd2bbf96 : SUCCESS in 5m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fff1245047674f6c9e3fc62bb3773baf : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"563789c2a378330838c700af2b750745c4d00c01","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-03 15:01:18.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"98a1dc22082fb1e45b13829829d38d690b4b9abc","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-03-03 15:04:57.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"99495b286ac0c8c2ed928c8e171306fe74250fc5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-06 08:37:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"37bba35bada904ac88f7c8aebab24f3f6de646e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 08:38:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d4261ff0cb8e53f0daa73f4bc67d370c20e8f14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 09:03:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ba45fb96c7841c78a422efbe2e6883c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca2174f00bfa4f9c8880ac90aea59da3 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf04272c42743e8a061ff7568075a35 : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c7c5e684bab465385a74cfa93cdf1ec : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":3},{"id":"be6a72b4276fee210faca211fc59cbfa94d537de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 09:03:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3a7a3a50a4744c0f899e9fdd0d6fe880a3754ef3","revisions":{"8a97f25114450409eb4efd49130d48a8e0141256":{"kind":"REWORK","_number":1,"created":"2023-02-27 12:24:01.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/71/875471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/875471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/875471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/875471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/875471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/875471/1"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:23:58.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:23:58.000000000","tz":60},"subject":"Release tacker RC1 for Antelope","message":"Release tacker RC1 for Antelope\n\nThis proposed a RC1 release for tacker to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 8.0.0.0rc1..05a4a178\nf951a43c Fix servernotification metadata issue\n33ca6ad5 Support tacker-db-manage for Multi DB backend\n4e699aec OAuth 2.0 Mutual-TLS Support\n2bc6d4d2 Fix image handling of StandardUserData heal\n4f371b86 Enable vnfcInfo id to be fixed\ndbc2caac Fix error attribute of LCM notification\n6e72a2d3 Add jammy support for tacker installer\n25459c25 Placement enhancement enables to AZ reselection\n2d5f6382 Update vimConnectionInfo\u0027s k8s parameter in v2\n7d7d6f1d Fix oidc\u0027s vim not applied to FT\nda52c8d0 Update k8s v1.25.6 and helm 3.10.3\n9aa5f0bc Fix missing teardown in PrometheusPluginTest\n5acf25a7 change_vnfpkg enhance to enable network change\nc592c0c2 Add VNF package content cache\nb74c0333 Use new get_rpc_client API from oslo.messaging\nc5b0806c Fix rollback error caused by DB sync\nffdbefae Stabilize sol-kubernetes-v2 job\n6720e274 Run Tacker gate job in Ubuntu Focal\n6f06cd42 Workaround for CI failure due to Heat timeout\n68b8a71c Restore overwritten documents\nd524b446 Support isAutohealEnabled for faultnotification\n68ff7a67 Fix missing rpc.init for mgmt driver\ne3e3fc51 Deprecate Legacy APIs excluding VIM feature\n3ad1976b Add API ref to tacker doc index\n01af9216 Switch the Legacy FT non-voting\n5b02232a Prometheus rule file syntax test\n7bb1bf8f Fix tox4 error\n111d1554 Migrate Tacker jobs to Ubuntu Jammy (22.04)\n995f01ba Update python classifier for python 3.10\n2f89ccc8 Fixed some bugs in individual VNFc management\ne491c60e Fix prometheus disk usage monitoring\n6ca63d6f Add unqualified search registries for k8s testing\nbe6d22b4 Fix waiting for pod removal when container update\nc297b430 Improve plugin flexibility\n58fbcc37 Fix field of LifecycleChangeNotificationsFilter\n3a1ccca9 Update k8s v1.25.2 and helm 3.10.1\n97bbbd62 Fix instantiate failed after terminate operation\nc1c9c907 Imported Translations from Zanata\n3ae06725 Stabilize sol-kubernetes-v2 job in Zuul FT\n594fffcc CNF Deployment failed due to missing Default VIM\nec835c33 Imported Translations from Zanata\nccf87eed CI: Revert temporary fixes for aodh and ceilometer\nd8edcf11 CI: Revert temporary changes for Z release RC\n77ecc2a7 Fix Vagrantfile for no_proxy is empty\n2e92de26 Fix k8s resource name handling in CNF v2 API\n4737a70a Fix Prometheus Plugin syntax error\n190e69cb A DB migration tool for API version update\na4f1052d Switch to 2023.1 Python3 unit tests and generic template name\n1e147a3c Update master for stable/zed\n0e2434d9 Database sync between TackerDB and Kubernetes\n02d947e3 Support multi artifact of interfaces\n03dcde32 Support multi artifact of ansible driver\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ic1f77c651d1f7c315f86d213561878dba1e58e57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a97f25114450409eb4efd49130d48a8e0141256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a97f25114450409eb4efd49130d48a8e0141256"}]},"branch":"refs/heads/master"},"a91d6f9f87ecfee817f2ca763d695278a8c24c29":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:00:54.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/71/875471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/875471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/875471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/875471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/875471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/875471/2"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:23:58.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 14:00:54.000000000","tz":0},"subject":"Release tacker RC1 for Antelope","message":"Release tacker RC1 for Antelope\n\nThis proposed a RC1 release for tacker to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 8.0.0.0rc1..05a4a178\nf951a43c Fix servernotification metadata issue\n33ca6ad5 Support tacker-db-manage for Multi DB backend\n4e699aec OAuth 2.0 Mutual-TLS Support\n2bc6d4d2 Fix image handling of StandardUserData heal\n4f371b86 Enable vnfcInfo id to be fixed\ndbc2caac Fix error attribute of LCM notification\n6e72a2d3 Add jammy support for tacker installer\n25459c25 Placement enhancement enables to AZ reselection\n2d5f6382 Update vimConnectionInfo\u0027s k8s parameter in v2\n7d7d6f1d Fix oidc\u0027s vim not applied to FT\nda52c8d0 Update k8s v1.25.6 and helm 3.10.3\n9aa5f0bc Fix missing teardown in PrometheusPluginTest\n5acf25a7 change_vnfpkg enhance to enable network change\nc592c0c2 Add VNF package content cache\nb74c0333 Use new get_rpc_client API from oslo.messaging\nc5b0806c Fix rollback error caused by DB sync\nffdbefae Stabilize sol-kubernetes-v2 job\n6720e274 Run Tacker gate job in Ubuntu Focal\n6f06cd42 Workaround for CI failure due to Heat timeout\n68b8a71c Restore overwritten documents\nd524b446 Support isAutohealEnabled for faultnotification\n68ff7a67 Fix missing rpc.init for mgmt driver\ne3e3fc51 Deprecate Legacy APIs excluding VIM feature\n3ad1976b Add API ref to tacker doc index\n01af9216 Switch the Legacy FT non-voting\n5b02232a Prometheus rule file syntax test\n7bb1bf8f Fix tox4 error\n111d1554 Migrate Tacker jobs to Ubuntu Jammy (22.04)\n995f01ba Update python classifier for python 3.10\n2f89ccc8 Fixed some bugs in individual VNFc management\ne491c60e Fix prometheus disk usage monitoring\n6ca63d6f Add unqualified search registries for k8s testing\nbe6d22b4 Fix waiting for pod removal when container update\nc297b430 Improve plugin flexibility\n58fbcc37 Fix field of LifecycleChangeNotificationsFilter\n3a1ccca9 Update k8s v1.25.2 and helm 3.10.1\n97bbbd62 Fix instantiate failed after terminate operation\nc1c9c907 Imported Translations from Zanata\n3ae06725 Stabilize sol-kubernetes-v2 job in Zuul FT\n594fffcc CNF Deployment failed due to missing Default VIM\nec835c33 Imported Translations from Zanata\nccf87eed CI: Revert temporary fixes for aodh and ceilometer\nd8edcf11 CI: Revert temporary changes for Z release RC\n77ecc2a7 Fix Vagrantfile for no_proxy is empty\n2e92de26 Fix k8s resource name handling in CNF v2 API\n4737a70a Fix Prometheus Plugin syntax error\n190e69cb A DB migration tool for API version update\na4f1052d Switch to 2023.1 Python3 unit tests and generic template name\n1e147a3c Update master for stable/zed\n0e2434d9 Database sync between TackerDB and Kubernetes\n02d947e3 Support multi artifact of interfaces\n03dcde32 Support multi artifact of ansible driver\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ic1f77c651d1f7c315f86d213561878dba1e58e57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a91d6f9f87ecfee817f2ca763d695278a8c24c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a91d6f9f87ecfee817f2ca763d695278a8c24c29"}]},"branch":"refs/heads/master"},"3a7a3a50a4744c0f899e9fdd0d6fe880a3754ef3":{"kind":"REWORK","_number":3,"created":"2023-03-03 14:42:46.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/71/875471/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/875471/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/875471/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/875471/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/875471/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/875471/3"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:23:58.000000000","tz":60},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2023-03-03 14:36:02.000000000","tz":0},"subject":"Release tacker RC1 for Antelope","message":"Release tacker RC1 for Antelope\n\nThis proposed a RC1 release for tacker to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 8.0.0.0rc1..af88b178\n9e49fc26 Add prelude section for 2023.1 releasenote\n74adf22d Add user docs for Coordination API sample usage\nc004eb25 Add docs for auto scale via PM Threshold API\n403c0ab4 Call heal interface changed to function\n5ab59f7e Support PromQL config file for PrometheusPlugin\naac03cef Enhancement of HTTP Client\n96d4ac63 Add support VNFM auto heal and scale\n0ff80687 Add doc of OAuth 2.0 Mutual-TLS Support\na423efca Support VNF LCM Coordination IF in change_vnfpkg\nc0620728 Missing Location in http header\n321da73b Add ChangeCurrentVnfPackage doc to user guide\n4e074e92 Add docs of VNFM auto heal and scale\n94d2433d Fix to upload as the appropriate image format\n2cbd84c5 Extend vim register function for helm in v2 API\nbb16f782 Update docs of User guide and API reference\n7f5c3906 Fix incorrect lcmocc_id specified in \"get_lcmocc\"\ned0ba564 Remove VIM monitoring depending on Mistral\nf951a43c Fix servernotification metadata issue\n3ea490ea Add Placement error-handling to user guide\n33ca6ad5 Support tacker-db-manage for Multi DB backend\n4e699aec OAuth 2.0 Mutual-TLS Support\n2bc6d4d2 Fix image handling of StandardUserData heal\n4f371b86 Enable vnfcInfo id to be fixed\ndbc2caac Fix error attribute of LCM notification\n6e72a2d3 Add jammy support for tacker installer\n25459c25 Placement enhancement enables to AZ reselection\n2d5f6382 Update vimConnectionInfo\u0027s k8s parameter in v2\n7d7d6f1d Fix oidc\u0027s vim not applied to FT\nda52c8d0 Update k8s v1.25.6 and helm 3.10.3\n9aa5f0bc Fix missing teardown in PrometheusPluginTest\n5acf25a7 change_vnfpkg enhance to enable network change\nc592c0c2 Add VNF package content cache\nb74c0333 Use new get_rpc_client API from oslo.messaging\nc5b0806c Fix rollback error caused by DB sync\nffdbefae Stabilize sol-kubernetes-v2 job\n6720e274 Run Tacker gate job in Ubuntu Focal\n6f06cd42 Workaround for CI failure due to Heat timeout\n68b8a71c Restore overwritten documents\nd524b446 Support isAutohealEnabled for faultnotification\n68ff7a67 Fix missing rpc.init for mgmt driver\ne3e3fc51 Deprecate Legacy APIs excluding VIM feature\n3ad1976b Add API ref to tacker doc index\n01af9216 Switch the Legacy FT non-voting\n5b02232a Prometheus rule file syntax test\n7bb1bf8f Fix tox4 error\n111d1554 Migrate Tacker jobs to Ubuntu Jammy (22.04)\n995f01ba Update python classifier for python 3.10\n2f89ccc8 Fixed some bugs in individual VNFc management\ne491c60e Fix prometheus disk usage monitoring\n6ca63d6f Add unqualified search registries for k8s testing\nbe6d22b4 Fix waiting for pod removal when container update\nc297b430 Improve plugin flexibility\n58fbcc37 Fix field of LifecycleChangeNotificationsFilter\n3a1ccca9 Update k8s v1.25.2 and helm 3.10.1\n97bbbd62 Fix instantiate failed after terminate operation\nc1c9c907 Imported Translations from Zanata\n3ae06725 Stabilize sol-kubernetes-v2 job in Zuul FT\n594fffcc CNF Deployment failed due to missing Default VIM\nec835c33 Imported Translations from Zanata\nccf87eed CI: Revert temporary fixes for aodh and ceilometer\nd8edcf11 CI: Revert temporary changes for Z release RC\n77ecc2a7 Fix Vagrantfile for no_proxy is empty\n2e92de26 Fix k8s resource name handling in CNF v2 API\n4737a70a Fix Prometheus Plugin syntax error\n190e69cb A DB migration tool for API version update\na4f1052d Switch to 2023.1 Python3 unit tests and generic template name\n1e147a3c Update master for stable/zed\n0e2434d9 Database sync between TackerDB and Kubernetes\n02d947e3 Support multi artifact of interfaces\n03dcde32 Support multi artifact of ansible driver\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nSigned-off-by: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nChange-Id: Ic1f77c651d1f7c315f86d213561878dba1e58e57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3a7a3a50a4744c0f899e9fdd0d6fe880a3754ef3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3a7a3a50a4744c0f899e9fdd0d6fe880a3754ef3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
