)]}'
{"id":"openstack%2Freleases~875476","triplet_id":"openstack%2Freleases~master~I36c91c9715e309763a86994ba6b53337b33ea05a","project":"openstack/releases","branch":"master","topic":"antelope-rc1-deadline","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-03-06 09:04:10.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-06 09:04:10.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2023-03-06 09:04:10.000000000","reason":"Change was submitted"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-03-06 09:04:10.000000000","reason":"Change was submitted"},"24061":{"account":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"last_update":"2023-03-06 09:04:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36c91c9715e309763a86994ba6b53337b33ea05a","subject":"Release venus RC1 for Antelope","status":"MERGED","created":"2023-02-27 12:26:59.000000000","updated":"2023-03-06 09:04:10.000000000","submitted":"2023-03-06 09:04:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875476-antelope-rc1-deadline","meta_rev_id":"013ca7d8a45394ec37c5d062c34dd082b104cec1","_number":875476,"virtual_id_number":875476,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-06 09:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-03-06 08:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-03-06 08:41:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 12:28:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 13:03:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 13:20:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"state":"REVIEWER"},{"updated":"2023-02-27 13:20:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2023-02-27 13:20:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2023-03-06 08:41:23.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"3e5f269fdc333547770ce7af72b7b725b12629af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 12:26:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db83b5612603b3124a6caaffd0bf57c416423ae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 13:03:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/481ba7734dd34affb3a6da69343d9e5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd19b196fb745ea9dd074b4bb398019 : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78059f3dad3a4640887e3357087ca039 : SUCCESS in 5m 10s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/291e3209301a45efa56cd6fd8ef38f06 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"bc5c8cf8026a6c7aab41e82d334c828b37128fdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:09:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ea19090e4b78343a25c5030e32754652c2fffcb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:16:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/254c8f1322da4c4ba8479439d52e41f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af370b0c63b7466c9a942492b0d48d30 : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41140a8a7810490c9b63101f4db9a1cf : SUCCESS in 5m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e0324b928da94a738118ca7879bc2c3f : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"15fa9280f1d96df2f9d8b4557142102a9d9f4916","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-03 10:42:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5d58b4fb0e691662a5978e170fa09f8459bf8b9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-03 13:14:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a9dfeb782610e7f6b4abae1f40c4cead2d377ab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 13:40:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56313f7fc6da485bb8845d0a888510a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3282ad06a18a401786d872396f28a02e : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c92e6ad0c33e407abfde9d0ccc9188d8 : SUCCESS in 5m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/04ccc13facdf4db996cfe2e78b43716e : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"84bae9c9782083fc55ee07ad3d678d5f8622f186","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-06 08:41:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4ab0c1e3bc56fed9d5ebff9db95820e774a600f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 08:42:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"828fa4893a8c803495576815a6673bb91d8f0d61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 09:04:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6852009ab594c5a8317cf9b7d4b52b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ba3134730346b981c6ac22713705f4 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740978a1ab674983829f3aabbeaad879 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/acd96a21532c4d4a85a1ac6e3bc4504b : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"013ca7d8a45394ec37c5d062c34dd082b104cec1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 09:04:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2d9738aca3c5a45edaea7bd41dc9cbb6a2513c1f","revisions":{"9a8cec03a37a75810161a914ec68eb5955a60b9c":{"kind":"REWORK","_number":1,"created":"2023-02-27 12:26:59.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/875476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/875476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/875476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/875476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/875476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/875476/1"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:26:55.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:26:55.000000000","tz":60},"subject":"Release venus RC1 for Antelope","message":"Release venus RC1 for Antelope\n\nThis proposed a RC1 release for venus to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 1.0.0..71a1ace\n71a1ace detail fixed\ncb884ce improve the desc of apis\ne86104e delete obsolete files\n390a59d adjust api document structure\na8e8fad add set_costom_config api ref\nd360c61 add api:set custom config\n8b6dc69 rename custom config example file\na951baa remove api outdated description\n78d1871 all index output to log\nbf49e28 log output after judge it\n36b3e34 set check delete es index time as 120s\n06e626f Fix tox4 error for venus\na1b6770 modify error word secion\n47e4e16 setting task execute interval as 60s\nd477ae1 hacking: force explicit import of python\u0027s mock\necafa92 add try-except when request a remote url\n8206ba1 update delete task config from db\ne90b293 fixbug:request can\u0027t get response\n64b8dbc remove duplicated function get_session\n58d4111 remove duplicated function\n3a61671 add databases modules\n8164828 add code of venusdatebase\n41a116e Add devstack steps.\nc81336f Supplement a test case for instance_call_chain.\n4c33371 Supplement a test case \u0027no_index\u0027 for typical_logs.\n16b0cbd Add a test case for search_typical_logs.\n2624f6d Supplement a unit test case for \u0027typical_logs\u0027.\n79e14f3 Supplement a test case \u0027no_index\u0027.\n87c2282 Add a unit test case to search_analyse_logs.\nab0d345 Add a unit test case \u0027invalid_index_type\u0027.\n9adc236 Supplement a unit test case \u0027valid_index_type\u0027.\n8abfaa3 Add a test case to search_logs.\n3f5cd82 Add a test case \u0027valid type\u0027 to search_params.\nddd8761 Add A unit test case to search_params.\n16ac2c9 Add unit test for controller api\n93184e1 Add fake request for test\neab47a9 remove tenant\n665230e Fix pbr version check\n83b1fb7 Add default conf for es\nc193e7e Add a use case for the test unit\n94fe8fb Supplement test case for logs\n9274f5f Update master for stable/zed\n6a797aa Switch to 2023.1 Python3 unit tests and generic template name\n7de7d89 fix details of search-instance-callchain.inc\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I36c91c9715e309763a86994ba6b53337b33ea05a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9a8cec03a37a75810161a914ec68eb5955a60b9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9a8cec03a37a75810161a914ec68eb5955a60b9c"}]},"branch":"refs/heads/master"},"1083ed0166ac9db1c229f031c8fe061d50d1d873":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:09:09.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/875476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/875476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/875476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/875476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/875476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/875476/2"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:26:55.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-27 14:09:09.000000000","tz":0},"subject":"Release venus RC1 for Antelope","message":"Release venus RC1 for Antelope\n\nThis proposed a RC1 release for venus to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\n$ git log --oneline --no-merges 1.0.0..71a1ace\n71a1ace detail fixed\ncb884ce improve the desc of apis\ne86104e delete obsolete files\n390a59d adjust api document structure\na8e8fad add set_costom_config api ref\nd360c61 add api:set custom config\n8b6dc69 rename custom config example file\na951baa remove api outdated description\n78d1871 all index output to log\nbf49e28 log output after judge it\n36b3e34 set check delete es index time as 120s\n06e626f Fix tox4 error for venus\na1b6770 modify error word secion\n47e4e16 setting task execute interval as 60s\nd477ae1 hacking: force explicit import of python\u0027s mock\necafa92 add try-except when request a remote url\n8206ba1 update delete task config from db\ne90b293 fixbug:request can\u0027t get response\n64b8dbc remove duplicated function get_session\n58d4111 remove duplicated function\n3a61671 add databases modules\n8164828 add code of venusdatebase\n41a116e Add devstack steps.\nc81336f Supplement a test case for instance_call_chain.\n4c33371 Supplement a test case \u0027no_index\u0027 for typical_logs.\n16b0cbd Add a test case for search_typical_logs.\n2624f6d Supplement a unit test case for \u0027typical_logs\u0027.\n79e14f3 Supplement a test case \u0027no_index\u0027.\n87c2282 Add a unit test case to search_analyse_logs.\nab0d345 Add a unit test case \u0027invalid_index_type\u0027.\n9adc236 Supplement a unit test case \u0027valid_index_type\u0027.\n8abfaa3 Add a test case to search_logs.\n3f5cd82 Add a test case \u0027valid type\u0027 to search_params.\nddd8761 Add A unit test case to search_params.\n16ac2c9 Add unit test for controller api\n93184e1 Add fake request for test\neab47a9 remove tenant\n665230e Fix pbr version check\n83b1fb7 Add default conf for es\nc193e7e Add a use case for the test unit\n94fe8fb Supplement test case for logs\n9274f5f Update master for stable/zed\n6a797aa Switch to 2023.1 Python3 unit tests and generic template name\n7de7d89 fix details of search-instance-callchain.inc\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I36c91c9715e309763a86994ba6b53337b33ea05a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1083ed0166ac9db1c229f031c8fe061d50d1d873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1083ed0166ac9db1c229f031c8fe061d50d1d873"}]},"branch":"refs/heads/master"},"2d9738aca3c5a45edaea7bd41dc9cbb6a2513c1f":{"kind":"REWORK","_number":3,"created":"2023-03-03 13:14:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/76/875476/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/875476/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/875476/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/875476/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/875476/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/875476/3"}}},"commit":{"parents":[{"commit":"e083e3051291103f970bd839e053b19be7434e67","subject":"Merge \"Release patrole for 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e083e3051291103f970bd839e053b19be7434e67"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-27 12:26:55.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-03-03 13:13:42.000000000","tz":60},"subject":"Release venus RC1 for Antelope","message":"Release venus RC1 for Antelope\n\nThis proposed a RC1 release for venus to get any merged changes\nout prior to the RC1 deadline for Antelope going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, March 03.\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I36c91c9715e309763a86994ba6b53337b33ea05a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d9738aca3c5a45edaea7bd41dc9cbb6a2513c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d9738aca3c5a45edaea7bd41dc9cbb6a2513c1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
