)]}'
{"id":"openstack%2Freleases~876951","triplet_id":"openstack%2Freleases~master~Idd42a387c35a34dba5da977027d21749a7884f88","project":"openstack/releases","branch":"master","topic":"fix-releasenotes-script","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-03-09 17:21:02.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-09 17:21:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd42a387c35a34dba5da977027d21749a7884f88","subject":"Allow to pass release ID to generate correct releasenotes links","status":"MERGED","created":"2023-03-09 10:27:07.000000000","updated":"2023-03-09 17:21:02.000000000","submitted":"2023-03-09 17:21:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"876951-fix-releasenotes-script","meta_rev_id":"825ad5b752ebeb6a874dec5815f67e3e6c44eb46","_number":876951,"virtual_id_number":876951,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-09 17:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-03-09 13:30:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-03-09 11:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-03-09 13:30:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-09 10:33:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-03-09 11:23:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 13:30:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"dce157edce2a606000a73190049574c50bbcc12a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-09 10:27:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e7fbb8207c66940e5e4e69e79667bfec5f25245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-09 10:27:42.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4625ca0a7c5194e0f197a0d6de72ab51417c4bc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-09 10:32:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4314dffa24715a29ed375e0018628861c19bc868","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-09 10:33:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ce46e17309066b65df3d9c2334c084c76d39dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-09 10:36:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"219517ad2804c234d33da0124a7e1956d17355ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-09 10:40:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"750ffe831d201810a4170e5a435fd7d6b2ee33e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-09 10:41:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1a28bd818549b10d42583d11660253bd6e64d0d8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-09 10:49:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4b13dd6915c49c4ffe60fe4224ac9e2d76096873","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-09 10:50:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"49a5b049d106a9180e63103f8f18b20f78c1d114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-09 10:54:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ce0d02292278dce6610d0d89e15ff8122d1c712f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-09 10:56:46.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"06f8dbbff145355f597ba4dbb8deb038915bc78b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-09 10:59:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"22264a7cb66e0a9bd27d86d362e3d678f7a9aee1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-09 11:05:00.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"648dcbad482b51840343a38b76853756b6162608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 11:23:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0607397ef88f474eaf68b2eb30393669\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ea4198df45464d9f5d9d8d8c103009 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c0b01070df642ab96ee4c21d2833037 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a31b0591af841a58e658d75d04e2653 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d788a7f59278483c922986590af22251 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf7d6b5d04f4c18bfc5e693ff4f6b19 : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8744a6ab9d2b4b75acd75fd6859cba0c : SUCCESS in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cf4747bbea5d42069502b4e453e1f90b : SUCCESS in 3m 31s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1c906da4397e434d88bbea07ac5a9db2 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":8},{"id":"287fdc0fb6243721813354138d792d25f6a7ccc8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-09 13:30:01.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"64cd5a3ee85659c2ea8b7152d1d23f5704a5f6ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 13:30:33.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6c80545aaa00622f35dcd573aad685998327e15d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 13:55:17.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19d09f60534949bfa5395e348fba2a6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef49759f4cec416da416343d5d0c146a : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b644041399ea46039952139a03e5d52d : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c12aeeb0ae5c487e8268276e4b48b4ff : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee7f7745e7e49e598f0337d23ce79e7 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fcf91529d8e046809600da641e232478 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbcba19a5d0f48158ded1215ae697b9f : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":8},{"id":"472d3394273093141f4f8a23b2e1774fb4ac7ea6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 17:08:08.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"dd3106923acc5f9672b4c6977fb1538c8d6c8941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 17:21:01.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbcb5fb6b7c347b5bd9729ec52bb527d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c2ba5131864640949e01fb5d30857b : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f01bbadc62e4a3a921c3f976f8c554f : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12495dae1f314524a2e14f3e8f295ac1 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e535de07d7ed4d9b95ecb19f6b89d4df : SUCCESS in 7m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/770599424c6648ea88891d25fb69a654 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c880926651045be8ed2fd18341071e8 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":8},{"id":"825ad5b752ebeb6a874dec5815f67e3e6c44eb46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 17:21:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"92684446d535f020a3a8451404c64d5165aa9634","revisions":{"a412e431ad11dd7820e7cd8b14a34b5010a91905":{"kind":"REWORK","_number":1,"created":"2023-03-09 10:27:07.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/876951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/876951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/876951/1"}}},"commit":{"parents":[{"commit":"b04a794336ee858b337882b38978ff285916ea46","subject":"Merge \"Release freezer-tempest-plugin for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b04a794336ee858b337882b38978ff285916ea46"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:26:08.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:26:08.000000000","tz":60},"subject":"allow to series\u0027s ID to generate correct releasenotes links","message":"allow to series\u0027s ID to generate correct releasenotes links\n\nChange-Id: Idd42a387c35a34dba5da977027d21749a7884f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a412e431ad11dd7820e7cd8b14a34b5010a91905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a412e431ad11dd7820e7cd8b14a34b5010a91905"}]},"branch":"refs/heads/master"},"d8c31fe4ff3e8a4c00c99bab47dabfe1fb79b1bf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-09 10:27:42.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/876951/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/876951/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/876951/2"}}},"commit":{"parents":[{"commit":"b04a794336ee858b337882b38978ff285916ea46","subject":"Merge \"Release freezer-tempest-plugin for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b04a794336ee858b337882b38978ff285916ea46"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:26:08.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:27:31.000000000","tz":60},"subject":"allow to pass the series\u0027s ID to generate correct releasenotes links","message":"allow to pass the series\u0027s ID to generate correct releasenotes links\n\nChange-Id: Idd42a387c35a34dba5da977027d21749a7884f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d8c31fe4ff3e8a4c00c99bab47dabfe1fb79b1bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d8c31fe4ff3e8a4c00c99bab47dabfe1fb79b1bf"}]},"branch":"refs/heads/master"},"1a8ab0e837705e644203bfc8161c5466bbfb44f8":{"kind":"REWORK","_number":3,"created":"2023-03-09 10:32:56.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/876951/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/876951/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/876951/3"}}},"commit":{"parents":[{"commit":"b04a794336ee858b337882b38978ff285916ea46","subject":"Merge \"Release freezer-tempest-plugin for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b04a794336ee858b337882b38978ff285916ea46"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:26:08.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:32:49.000000000","tz":60},"subject":"allow to pass the series\u0027s ID to generate correct releasenotes links","message":"allow to pass the series\u0027s ID to generate correct releasenotes links\n\nChange-Id: Idd42a387c35a34dba5da977027d21749a7884f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a8ab0e837705e644203bfc8161c5466bbfb44f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a8ab0e837705e644203bfc8161c5466bbfb44f8"}]},"branch":"refs/heads/master"},"b1dfac1b05b62d53115105be40f54c7e9e554978":{"kind":"REWORK","_number":4,"created":"2023-03-09 10:36:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/876951/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/876951/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/876951/4"}}},"commit":{"parents":[{"commit":"b04a794336ee858b337882b38978ff285916ea46","subject":"Merge \"Release freezer-tempest-plugin for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b04a794336ee858b337882b38978ff285916ea46"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:26:08.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:36:40.000000000","tz":60},"subject":"allow to pass the series\u0027s ID to generate correct releasenotes links","message":"allow to pass the series\u0027s ID to generate correct releasenotes links\n\nChange-Id: Idd42a387c35a34dba5da977027d21749a7884f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1dfac1b05b62d53115105be40f54c7e9e554978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1dfac1b05b62d53115105be40f54c7e9e554978"}]},"branch":"refs/heads/master"},"70b2fc573641f4e1e1ac397d1502bf45b34b7fbc":{"kind":"REWORK","_number":5,"created":"2023-03-09 10:40:02.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/876951/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/876951/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/876951/5"}}},"commit":{"parents":[{"commit":"b04a794336ee858b337882b38978ff285916ea46","subject":"Merge \"Release freezer-tempest-plugin for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b04a794336ee858b337882b38978ff285916ea46"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:26:08.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:39:55.000000000","tz":60},"subject":"allow to pass the series\u0027s ID to generate correct releasenotes links","message":"allow to pass the series\u0027s ID to generate correct releasenotes links\n\nChange-Id: Idd42a387c35a34dba5da977027d21749a7884f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/70b2fc573641f4e1e1ac397d1502bf45b34b7fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/70b2fc573641f4e1e1ac397d1502bf45b34b7fbc"}]},"branch":"refs/heads/master"},"52ce86c2243b78013c1f75c9d4463915c26ffd91":{"kind":"REWORK","_number":6,"created":"2023-03-09 10:41:23.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/876951/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/876951/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/876951/6"}}},"commit":{"parents":[{"commit":"b04a794336ee858b337882b38978ff285916ea46","subject":"Merge \"Release freezer-tempest-plugin for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b04a794336ee858b337882b38978ff285916ea46"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:26:08.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:41:18.000000000","tz":60},"subject":"allow to pass the series\u0027s ID to generate correct releasenotes links","message":"allow to pass the series\u0027s ID to generate correct releasenotes links\n\nChange-Id: Idd42a387c35a34dba5da977027d21749a7884f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/52ce86c2243b78013c1f75c9d4463915c26ffd91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/52ce86c2243b78013c1f75c9d4463915c26ffd91"}]},"branch":"refs/heads/master"},"e338207ee311699e4e460e7956bcd5bd05099238":{"kind":"REWORK","_number":7,"created":"2023-03-09 10:54:33.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/51/876951/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/876951/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/876951/7"}}},"commit":{"parents":[{"commit":"b04a794336ee858b337882b38978ff285916ea46","subject":"Merge \"Release freezer-tempest-plugin for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b04a794336ee858b337882b38978ff285916ea46"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:26:08.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-03-09 10:52:21.000000000","tz":60},"subject":"Allow to pass release ID to generate correct releasenotes links","message":"Allow to pass release ID to generate correct releasenotes links\n\nChange-Id: Idd42a387c35a34dba5da977027d21749a7884f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e338207ee311699e4e460e7956bcd5bd05099238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e338207ee311699e4e460e7956bcd5bd05099238"}]},"branch":"refs/heads/master"},"92684446d535f020a3a8451404c64d5165aa9634":{"kind":"REWORK","_number":8,"created":"2023-03-09 10:59:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/51/876951/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/876951/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/876951/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/876951/8"}}},"commit":{"parents":[{"commit":"b04a794336ee858b337882b38978ff285916ea46","subject":"Merge \"Release freezer-tempest-plugin for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b04a794336ee858b337882b38978ff285916ea46"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-09 10:26:08.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-03-09 10:58:27.000000000","tz":60},"subject":"Allow to pass release ID to generate correct releasenotes links","message":"Allow to pass release ID to generate correct releasenotes links\n\nChange-Id: Idd42a387c35a34dba5da977027d21749a7884f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92684446d535f020a3a8451404c64d5165aa9634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92684446d535f020a3a8451404c64d5165aa9634"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
