)]}'
{"id":"openstack%2Freleases~877094","triplet_id":"openstack%2Freleases~master~I0f8c4c4a0990966e54a2c934e055bfcc2114ee17","project":"openstack/releases","branch":"master","topic":"nova_bobcat_deadlines","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2023-04-13 10:09:47.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-04-11 16:57:10.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-04-13 11:09:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f8c4c4a0990966e54a2c934e055bfcc2114ee17","subject":"Propose Nova deadlines for Bobcat schedule","status":"MERGED","created":"2023-03-10 12:00:57.000000000","updated":"2023-04-13 11:09:01.000000000","submitted":"2023-04-13 11:09:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877094-nova_bobcat_deadlines","meta_rev_id":"bffe6946ed23ec9523cb80f800a2f4a78a736d42","_number":877094,"virtual_id_number":877094,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 11:09:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-04-13 10:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-04-11 16:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-04-13 10:09:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-04-13 10:09:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-10 12:08:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-10 12:34:34.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-04-06 07:42:08.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-04-11 10:39:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"ee290a4d612060f76d2aec410007409bca43a261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-03-10 12:00:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a944f566761d711cab1cec14a23d37b25b1344a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 12:08:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cfe0454a1d548248c98844ef3ff9e9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b07ce2cfb24d3091676e8a437d560d : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a28ee9b1de4ab9ae4e1fcff2cf1613190d3eecba","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-10 12:34:34.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e6599316c5843c0742a84cb006496f15b5bd738a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-03-10 15:56:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12582ac463952e581d63d352b934d1bd424c5743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 16:09:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e96d98f02294ce7b3ef724aa3c27253\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b1a2d529eb4941a07803b2d8da52f3 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":2},{"id":"5d6fe8685f36a1e35fafff1f6027465fb39d40ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-04-04 14:18:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9ea06da2e2cc858167ede15480033723db9ee8f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 14:31:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01afd479de23465ba169ad01817f5d9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b8febf521c24f4fa75b33157389a22e : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":3},{"id":"5480aceb640f2c3f3cbc1dcce4f5c5821df88e76","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-06 07:42:08.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"bf02b6d2a70a299fbef19ff2acdfb46a5b939c54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-04-07 07:40:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:NO_CHANGE OR changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1 (copy condition: \"changekind:NO_CHANGE\")\n","accounts_in_message":[],"_revision_number":4},{"id":"5571b73fcc6df798505b64cb2be526f51dc4ad35","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-07 07:41:40.000000000","message":"Patch Set 4: Code-Review+2 PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"b77801d25f010f76fe73822a98f00c14b9792bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 07:52:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51dc6f43d6f5474f80d00a8aed617d53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b91e6d71c0c42dc83848349a89d1e4e : SUCCESS in 10m 24s","accounts_in_message":[],"_revision_number":4},{"id":"03d13970d9740bca34bcb8ac2964bdf5b113cbe1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-11 10:39:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"72e071273b903d8e7cc89cf70170f3f0719c48b5","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-04-11 16:53:51.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4f0b6351f686dee08d7ca090ad06fc3c7955524f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-04-11 16:54:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4cecea02dafc8e4fd28da7f093f17b250b42211d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-11 16:57:10.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"75591e056d3c8aef6072af60d2b7e44309052e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 17:00:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6abbe6a087d64a3385db6bed213a5c26\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be2776b6510046d7886c9b739c0131d6 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":5},{"id":"a75ff425e336d51deffb2ec8e4d4f820ee1e8642","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-13 10:09:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1 PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"4a80c582e4f4a421b845d3039a3a19c79b0b2ea9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 10:10:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"62e0aed9393f9e0a090693d48c80f951bc24f4eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 11:09:01.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76c5595a6b084fb9a043d48ec2b87c17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b73660c9dcb4146bef3c341eea15854 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa01c0a09a594237b22ceda15c4b0982 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":5},{"id":"bffe6946ed23ec9523cb80f800a2f4a78a736d42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 11:09:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"89ebab0d0d574d8ef2d71932068fba2016a5c545","revisions":{"2e3ffb76b466f2e0cad826a110eea21cece8ce4c":{"kind":"REWORK","_number":1,"created":"2023-03-10 12:00:57.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/94/877094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/877094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/877094/1"}}},"commit":{"parents":[{"commit":"6dcffc4b56c65c8a5cd1fcf88ceee153cf6aefb2","subject":"Merge \"Release swift 2.31.1 and cut branch stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dcffc4b56c65c8a5cd1fcf88ceee153cf6aefb2"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-10 12:00:00.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-10 12:00:00.000000000","tz":60},"subject":"WIP: Propose Nova deadlines for Bobcat schedule","message":"WIP: Propose Nova deadlines for Bobcat schedule\n\nWORK IN PROGRESS as the team needs to have consensus on those dates first.\n\nChange-Id: I0f8c4c4a0990966e54a2c934e055bfcc2114ee17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e3ffb76b466f2e0cad826a110eea21cece8ce4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e3ffb76b466f2e0cad826a110eea21cece8ce4c"}]},"branch":"refs/heads/master"},"3f999b964a6552a10ff84cbf3b3b4fc1961462ae":{"kind":"REWORK","_number":2,"created":"2023-03-10 15:56:15.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/94/877094/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/877094/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/877094/2"}}},"commit":{"parents":[{"commit":"6dcffc4b56c65c8a5cd1fcf88ceee153cf6aefb2","subject":"Merge \"Release swift 2.31.1 and cut branch stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dcffc4b56c65c8a5cd1fcf88ceee153cf6aefb2"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-10 12:00:00.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-10 15:56:08.000000000","tz":60},"subject":"WIP: Propose Nova deadlines for Bobcat schedule","message":"WIP: Propose Nova deadlines for Bobcat schedule\n\nWORK IN PROGRESS as the team needs to have consensus on those dates first.\n\nChange-Id: I0f8c4c4a0990966e54a2c934e055bfcc2114ee17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3f999b964a6552a10ff84cbf3b3b4fc1961462ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3f999b964a6552a10ff84cbf3b3b4fc1961462ae"}]},"branch":"refs/heads/master"},"5f5f384458841e43e23afe946006e43a10cb3644":{"kind":"REWORK","_number":3,"created":"2023-04-04 14:18:00.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/94/877094/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/877094/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/877094/3"}}},"commit":{"parents":[{"commit":"6dcffc4b56c65c8a5cd1fcf88ceee153cf6aefb2","subject":"Merge \"Release swift 2.31.1 and cut branch stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dcffc4b56c65c8a5cd1fcf88ceee153cf6aefb2"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-10 12:00:00.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-04-04 14:16:47.000000000","tz":120},"subject":"Propose Nova deadlines for Bobcat schedule","message":"Propose Nova deadlines for Bobcat schedule\n\nNova cores are encouraged to +1 or -1 this change until Wednesday Apr 5 1600UTC.\n\nChange-Id: I0f8c4c4a0990966e54a2c934e055bfcc2114ee17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f5f384458841e43e23afe946006e43a10cb3644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f5f384458841e43e23afe946006e43a10cb3644"}]},"branch":"refs/heads/master"},"089113fae826e639b6175518b1fa86c2f4e95146":{"kind":"REWORK","_number":4,"created":"2023-04-07 07:40:04.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/94/877094/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/877094/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/877094/4"}}},"commit":{"parents":[{"commit":"e28a06c78b4f8b71e978ae446d08f77e9b4107e3","subject":"Merge \"Add cinder specific dates for 2023.2 Bobcat Schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e28a06c78b4f8b71e978ae446d08f77e9b4107e3"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-10 12:00:00.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-04-07 07:39:52.000000000","tz":120},"subject":"Propose Nova deadlines for Bobcat schedule","message":"Propose Nova deadlines for Bobcat schedule\n\nNova cores are encouraged to +1 or -1 this change until Wednesday Apr 5 1600UTC.\n\nChange-Id: I0f8c4c4a0990966e54a2c934e055bfcc2114ee17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/089113fae826e639b6175518b1fa86c2f4e95146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/089113fae826e639b6175518b1fa86c2f4e95146"}]},"branch":"refs/heads/master"},"89ebab0d0d574d8ef2d71932068fba2016a5c545":{"kind":"REWORK","_number":5,"created":"2023-04-11 16:54:14.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/94/877094/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/877094/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/877094/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/877094/5"}}},"commit":{"parents":[{"commit":"e28a06c78b4f8b71e978ae446d08f77e9b4107e3","subject":"Merge \"Add cinder specific dates for 2023.2 Bobcat Schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e28a06c78b4f8b71e978ae446d08f77e9b4107e3"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-10 12:00:00.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-04-11 16:54:07.000000000","tz":120},"subject":"Propose Nova deadlines for Bobcat schedule","message":"Propose Nova deadlines for Bobcat schedule\n\nNova cores are encouraged to +1 or -1 this change until Wednesday Apr 5 1600UTC.\n\nChange-Id: I0f8c4c4a0990966e54a2c934e055bfcc2114ee17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89ebab0d0d574d8ef2d71932068fba2016a5c545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89ebab0d0d574d8ef2d71932068fba2016a5c545"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:PTL-Approved\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
