)]}'
{"id":"openstack%2Freleases~877105","triplet_id":"openstack%2Freleases~master~I014b1d6dc561be4db0fc8faa85fb3133e851acfc","project":"openstack/releases","branch":"master","topic":"release-id-as-branch-name","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-03-13 10:24:45.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-13 09:57:35.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I014b1d6dc561be4db0fc8faa85fb3133e851acfc","subject":"Fix gitutils to use release-id","status":"MERGED","created":"2023-03-10 15:15:46.000000000","updated":"2023-03-13 10:24:45.000000000","submitted":"2023-03-13 10:24:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877105-release-id-as-branch-name","meta_rev_id":"f3a0290eb34fc9a71027edb0a866fc2e8d7dec1c","_number":877105,"virtual_id_number":877105,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 10:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-03-13 09:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-03-13 09:57:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-03-13 09:57:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-10 15:27:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-13 09:54:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"d77412d88ebdc4c96762113f4d4f07a0e447fbea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-10 15:15:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdcacb1d6e10dbddfbcdc025cf3b6d96509de650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 15:27:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cb0fb7c0a8345eaade8e7fd41d33cb9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f17fd84a7e445a8dafd1c6f8f8497c : FAILURE in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0033feb7fff04ae5a2387e0c25e38fa8 : FAILURE in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d5eb3dce59c4dbb985315f05f1e5314 : FAILURE in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcd31472e46b4f86b49e883e35d3961c : FAILURE in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c5bb281319b4c0ab369252ced8f4179 : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb1c657a51cf4315a1ad6202d9313ac8 : SUCCESS in 3m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6c345b715251462181e142e85c3c0f81 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3ea8017e932dc758dda96e0ac1233f770beb9589","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-10 17:07:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5c9551addecab6b9959446bdbcf666944e4c3fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 17:17:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/429732ad16664f56bdfb9bd3802442de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/575bf208dd7d40e1b2074a512d25a908 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bce7d3b6c34c414d981518a2c9aa93e6 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8929eb7fba745e9a9b1c805ccc9721a : SUCCESS in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd68bdcedbf24f4aa3c30c190982eaef : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/182c1d1b08074fe587b2e37344afa925 : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8d5550be66c4137a89a2314c285191d : SUCCESS in 7m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4b3e5804fc1b4e5c8d065fdb3dc78a46 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b29010188bd23853a9920c28f7ef31e5cd153e11","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-13 09:54:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cd544b689d996ccd3c109f129b3309fac2012d35","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-13 09:57:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1cfbc8e130f89eb1dff1968a9651dea793c7ffb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 09:57:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac2c67081d9e31134438a4812d2dc0dc13436ef7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:24:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7be82ac9be7d46dfbecb6484422369e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20c479aa22054010915c23f041e652b8 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05f319f366ec45b3bc7ff8794180b444 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed947545d9a14e29adbcae17552012a6 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71a7f0410a6b4158b5da31b01deddc8a : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b4ffd440cd749d6b69dd2caac4488b9 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aeba79b436754e3f82412e57f1ed190c : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f3a0290eb34fc9a71027edb0a866fc2e8d7dec1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:24:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fb08afc1fa29664bf5e2abb6bd8c48cee87d8c5a","revisions":{"4681e525b10994d48ef204b4c24f45fe4ee22bca":{"kind":"REWORK","_number":1,"created":"2023-03-10 15:15:46.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/05/877105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/877105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/877105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/877105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/877105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/877105/1"}}},"commit":{"parents":[{"commit":"c957de22467ce41dcee04776f3e3b9c2710e53cc","subject":"Merge \"Release python-openstackclient 6.2.0 and create branch stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c957de22467ce41dcee04776f3e3b9c2710e53cc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 15:14:53.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 15:15:33.000000000","tz":60},"subject":"Fix gitutils to use release-id","message":"Fix gitutils to use release-id\n\nWith the new release identification / naming schema [1] (like:\n2023.1 Antelope) new stable branch naming was introduced (like:\nstable/2023.1). This was not handled in the list-deliverables command as\nit\nstill uses the release name. This patch fixes this by reading the\n\u0027release-id\u0027 field from series_status.yaml and if present then uses it\nas stable/\u003crelease-id\u003e for the branch creation.\n\n[1] https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: I014b1d6dc561be4db0fc8faa85fb3133e851acfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4681e525b10994d48ef204b4c24f45fe4ee22bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4681e525b10994d48ef204b4c24f45fe4ee22bca"}]},"branch":"refs/heads/master"},"fb08afc1fa29664bf5e2abb6bd8c48cee87d8c5a":{"kind":"REWORK","_number":2,"created":"2023-03-10 17:07:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/05/877105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/877105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/877105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/877105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/877105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/877105/2"}}},"commit":{"parents":[{"commit":"c957de22467ce41dcee04776f3e3b9c2710e53cc","subject":"Merge \"Release python-openstackclient 6.2.0 and create branch stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c957de22467ce41dcee04776f3e3b9c2710e53cc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 15:14:53.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-03-10 16:57:53.000000000","tz":60},"subject":"Fix gitutils to use release-id","message":"Fix gitutils to use release-id\n\nWith the new release identification / naming schema [1] (like:\n2023.1 Antelope) new stable branch naming was introduced (like:\nstable/2023.1). This was not handled in check_branch_sha method\nin validate.py, as the method still searches stable/$series.\n\nThis patch fixes this by reading the \u0027release-id\u0027 field from\nseries_status.yaml if present and uses it as stable/\u003crelease-id\u003e\nfor the branch search.\n\n[1] https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: I014b1d6dc561be4db0fc8faa85fb3133e851acfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb08afc1fa29664bf5e2abb6bd8c48cee87d8c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb08afc1fa29664bf5e2abb6bd8c48cee87d8c5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
