)]}'
{"id":"openstack%2Freleases~877819","triplet_id":"openstack%2Freleases~master~Ie8ba3e9bfa01148e93f5920050703e998248db11","project":"openstack/releases","branch":"master","topic":"fix-propose-final-releases","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-03-23 11:33:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie8ba3e9bfa01148e93f5920050703e998248db11","subject":"Fix propose-final-releases command to use release-id","status":"MERGED","created":"2023-03-17 15:51:50.000000000","updated":"2023-03-23 11:33:01.000000000","submitted":"2023-03-23 11:33:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"877819-fix-propose-final-releases","meta_rev_id":"042e31a356be265b3b7fad289c92528fd4122339","_number":877819,"virtual_id_number":877819,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-23 11:33:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-03-23 11:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-03-20 10:11:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-03-23 11:19:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 15:56:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-03-17 16:00:24.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-03-17 16:08:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 11:19:08.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"805fdaeb3f8d682d7bf0796180d38ea6166c6af7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-17 15:51:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef2a0f557ad417f3c81898ce75f1173b59cda539","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-17 15:56:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40da78dd5b64ed386e19225dfee2f3faa56728c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-17 15:58:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5a73e1f8154fc5b9c07b6324b72d00f51aac2eaf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-17 16:00:24.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"118238ebcb5ab459a7803d2858702533069c63fb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-17 16:01:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fb0f09d768a1d77eef2222e693f69bc4e942b55c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 16:08:28.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1786b7112cf4fb39ae0622bb9192756\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d8fb540d57e48d895a25e753f7a494b : FAILURE in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b76de233f6ba4454b7d1c64d76957e92 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0c01431416a48e7bb10ab0a541ebf63 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac5c24213528488a879f52dfa83a1aaa : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f935451b1c4b4495846f86aad8b88d : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5af6e2a355d54e68949c358f89845f6a : SUCCESS in 5m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d53fd4765b774d6f834b73a6bafea79d : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"5d0ee5d60a0dd0f4e7a4e51c4480c5efd183e5de","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-20 08:32:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48242aee803a5af6c98c681bd4286691c7da3251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 08:45:11.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2c5c1f3c92145978a776480df80da9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b11c66037e74db6a276b0f3d84c27d9 : FAILURE in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a65aafa542f5469c8b46c51696844b33 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aab1dd698c94a749c11cf16d575ab36 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/838106ccf4ca4bc4a0a80d4a311d542d : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b95e83d07f4ddbaf0ceb5cad2f23bc : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/137c284cab6044dfa9f69a23cd601a0b : SUCCESS in 4m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e666819fd25f46cd927daa4eefc86662 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"3b2b0a3c6345e5751c5bdca7378e7ce36ca8c292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-20 10:06:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"16f475d619effcef879f90f90bf9114e62aed8bb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-20 10:11:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cfd2f06f8f243367f9e96eb7dcb82576935250c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 10:16:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e92e15a76e0409aa94d935fcb19d75b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feda1403c3a44f17b78ae5c5ae74e949 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26650aa9a6b145cfb3080639adbd64e3 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcd268944f59460a86e0087f4f3c31c6 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f64fa64647234479b01cb32009d7151a : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce258a58adb4c099afcd98ca8ad4457 : SUCCESS in 6m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef55474e15564f55b9d0bfd2aca2769d : SUCCESS in 4m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/602908c09cbd45e28a000bb0c3265199 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"75f33789e9aec16e881beccd22c439b1841e17a4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-23 11:19:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a0049dd533873ceada795ae0ba5eb8e14a8939fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 11:19:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"238af3bcde331725e43bd3d659c2eee73c8213b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 11:33:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327cc933520d48b583b25b3046eb789f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a749da6ce57441c08e904d9d0c7a62c6 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0642e9579ba24287839a68e9f59371b6 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab110c81768e4899a581abca0f937eca : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68748af6ee0c4243ba033fa5fa575c70 : SUCCESS in 7m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75c16756c62b47c8a9b553ce1b08fa06 : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d36947083114888b58ca164c7369723 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"042e31a356be265b3b7fad289c92528fd4122339","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 11:33:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"14484c5055057cd22db07eba02081eea8d3962fc","revisions":{"11f3e4341ef3e6f1088ab0ade1d6cb30f7486d30":{"kind":"REWORK","_number":1,"created":"2023-03-17 15:51:50.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/19/877819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/877819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/877819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/877819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/877819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/877819/1"}}},"commit":{"parents":[{"commit":"06e3ca86bcd9e007642e0e010810991d16f8fc74","subject":"Merge \"Add missing release note links for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/06e3ca86bcd9e007642e0e010810991d16f8fc74"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-17 15:50:38.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-17 15:50:38.000000000","tz":60},"subject":"Fix propose-final-releases command to use release-id","message":"Fix propose-final-releases command to use release-id\n\nWith the new release identification / naming schema [1] (like:\n2023.1 Antelope) new stable branch naming was introduced (like:\nstable/2023.1). This was not handled in the list-deliverables command as it\nstill uses the release name. This patch fixes this by reading the\n\u0027release-id\u0027 field from series_status.yaml and if present then uses it\nas stable/\u003crelease-id\u003e for the branch creation.\n\n[1] https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: Ie8ba3e9bfa01148e93f5920050703e998248db11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11f3e4341ef3e6f1088ab0ade1d6cb30f7486d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11f3e4341ef3e6f1088ab0ade1d6cb30f7486d30"}]},"branch":"refs/heads/master"},"e4a15bd4765dc3582d1fcf15e0a0bf8e9a6e7e31":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-17 15:58:30.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/19/877819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/877819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/877819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/877819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/877819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/877819/2"}}},"commit":{"parents":[{"commit":"06e3ca86bcd9e007642e0e010810991d16f8fc74","subject":"Merge \"Add missing release note links for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/06e3ca86bcd9e007642e0e010810991d16f8fc74"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-17 15:50:38.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-17 15:58:01.000000000","tz":60},"subject":"Fix propose-final-releases command to use release-id","message":"Fix propose-final-releases command to use release-id\n\nWith the new release identification / naming schema [1] (like:\n2023.1 Antelope) new stable branch naming was introduced (like:\nstable/2023.1). This was not handled in the propose-final-releases command as\nit still uses the release name. This patch fixes this by reading the\n\u0027release-id\u0027 field from series_status.yaml and if present then uses it\nas stable/\u003crelease-id\u003e for the branch creation.\n\n[1] https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: Ie8ba3e9bfa01148e93f5920050703e998248db11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e4a15bd4765dc3582d1fcf15e0a0bf8e9a6e7e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e4a15bd4765dc3582d1fcf15e0a0bf8e9a6e7e31"}]},"branch":"refs/heads/master"},"14484c5055057cd22db07eba02081eea8d3962fc":{"kind":"REWORK","_number":3,"created":"2023-03-20 10:06:02.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/19/877819/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/877819/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/877819/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/877819/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/877819/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/877819/3"}}},"commit":{"parents":[{"commit":"06e3ca86bcd9e007642e0e010810991d16f8fc74","subject":"Merge \"Add missing release note links for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/06e3ca86bcd9e007642e0e010810991d16f8fc74"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-17 15:50:38.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-20 10:05:56.000000000","tz":60},"subject":"Fix propose-final-releases command to use release-id","message":"Fix propose-final-releases command to use release-id\n\nWith the new release identification / naming schema [1] (like:\n2023.1 Antelope) new stable branch naming was introduced (like:\nstable/2023.1). This was not handled in the propose-final-releases command as\nit still uses the release name. This patch fixes this by reading the\n\u0027release-id\u0027 field from series_status.yaml and if present then uses it\nas stable/\u003crelease-id\u003e for the branch creation.\n\n[1] https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: Ie8ba3e9bfa01148e93f5920050703e998248db11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14484c5055057cd22db07eba02081eea8d3962fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14484c5055057cd22db07eba02081eea8d3962fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:PTL-Approved\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
