)]}'
{"id":"openstack%2Freleases~878731","triplet_id":"openstack%2Freleases~master~I536d9cdd8933e3cc7aece42e129bd9d42b4bc0f6","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-05 15:25:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I536d9cdd8933e3cc7aece42e129bd9d42b4bc0f6","subject":"Puppet OpenStack 2023.1 release","status":"MERGED","created":"2023-03-28 02:16:21.000000000","updated":"2023-04-06 01:55:08.000000000","submitted":"2023-04-05 15:25:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878731","meta_rev_id":"06b97ad2f6b72689be04281e835b53976decb3c7","_number":878731,"virtual_id_number":878731,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-05 15:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-04-04 05:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-04-05 15:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-04-03 10:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-04-05 15:06:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-04-03 01:55:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 02:17:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-30 11:45:19.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-03-31 19:00:20.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-04-03 10:00:50.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"6cdb6103d9ce2be62f3bba17ecb9c07c525f0d2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-28 02:16:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd5feaa9f45b740f8169923fb56d219ff183eb8b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 02:17:35.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"d279751ef53131ce6fe80dd472593d3b4fdce99a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-28 02:17:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a254950168ccb38f31f0dbef8ddd0c68b76dc50a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-28 02:18:03.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"a64f27a103dad335b073d9399e0cf339f52de008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-28 02:18:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a293dfefe474e59c4f01fa674f2f25e9f2c45297","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-28 02:19:47.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"072008b5e70ce0e143319e2079a83906cdcefeab","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 02:21:07.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"24c58a9831d09e553dbe97fbe4d4d0a29103ad4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 03:11:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/600f5634f6d242448a34da971bfe3486\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b93505910274eaa8854142e62619a08 : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09acaf52edba4f60aea1839a8b7d262a : FAILURE in 14m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f5a31134f57a44d59d0c374f4eb7ca00 : FAILURE in 6m 35s","accounts_in_message":[],"_revision_number":3},{"id":"4a15f7a9bdd768273ce73feb059439fd52da9060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-28 14:01:33.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"e5e16d03d69f6180bdb5110d7be1972010bfb6e7","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 14:02:28.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"5164352a7b2898a9826d6303180a889d1053a625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 14:18:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edb2cac915b74e74a24e0fa4a42a646c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f24b2b530ec34ef2af76c8fc7313b1c6 : SUCCESS in 7m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c6c74053e554af69a5ed99e99c89945 : FAILURE in 15m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e57c002d58d841ddafbd9c3cce83099c : FAILURE in 6m 23s","accounts_in_message":[],"_revision_number":4},{"id":"13dcc53f1c1144b887121ebaac5f81a2e99acbc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-29 06:16:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"967953afa24e1549ec17cc2fd4ff258814772b74","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 06:17:43.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"b89fa681b9079c8833fe7c51a5458185be4ce6ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 06:37:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/019cfb1eb0214a79ad0b8eddea1b3ae9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7b27d2b65354a7eb4b59a7971368d5c : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd65aa0c079c45ea833036afe87639dd : FAILURE in 15m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/40755f1b1c6b414190591d64d116c9e0 : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":5},{"id":"39c6eb439e5d8ccc4b89f65e1edaa00ee546ad7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-29 06:43:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d512d59e7d2ce7244e592475044dd6c60e1309f4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 06:44:15.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"5f4c70a204f58e917b4564683f87ed0e0d202b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 07:07:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c54e864e4e640f797510ff24c6c1012\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22b70d055bc4bea87676c9f89557a0e : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b05aec4670be4c308f23a8d81ddfc3fe : SUCCESS in 18m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/13f027cd09854c5a9b9ddceab926066b : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":6},{"id":"96c369d2a5324155e10daee74dc3552e89953d29","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-30 11:45:19.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4f284f31873bafd3932f0290d5cc1cdb147388f8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-31 19:00:20.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"60e68e5e211631723af21a4528fd51f354f34201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-03 00:19:29.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"9fc55919ec84b74b53b552f22e25759b06153aab","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 00:20:20.000000000","message":"Patch Set 7: PTL-Approved+1","accounts_in_message":[],"_revision_number":7},{"id":"d35c7bdcf607f4e329e2fa8551f39785d2974e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 00:36:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4480e1112314bd78c37c32da82c9293\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/035bdc5770f9472baf6f347d9f861944 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a2076bcc18c41a9a1d46236d0911de7 : SUCCESS in 14m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/256470e0399b40b3b1c49ff234c92019 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":7},{"id":"79cbdb1d997f3e6015327235c4a93f830d37ef81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-03 01:54:39.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"27b7e902e24d29a5bfd46a50472342cb15e0828e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-03 01:55:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a8dce67ebd92fc195d3c9445f884b990000dbc3b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 01:55:27.000000000","message":"Patch Set 8: PTL-Approved+1","accounts_in_message":[],"_revision_number":8},{"id":"d60c41ba0d28160553652ce9a8adb97570e84732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 02:08:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/272ba09bcedb40acafe3b210a34d6c35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f546e3e56d437e801a1dacbd56af23 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a06614c791ff45958f07c2d22140ee54 : SUCCESS in 12m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9b93e1a31438417b999bdd6b11f5e0f8 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":8},{"id":"fb48c63f1e797dbb77224ff007ca7c9a7310e610","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-04-03 10:00:50.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"efa98775cb12d5904f1527281c736dd40cca2689","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-04 05:47:08.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"2f91e31a55ed61ac73ef1ec51744f4a03308c1af","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-05 15:06:00.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"702dd5aeec089a9e19319a3eb74edf8598504edf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 15:06:29.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"0ea7e58fbbcaa586cc2b10e87962870b213e794d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 15:25:18.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63162261e7774f7fa57c151b13fa8b81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b172eef592ef4ccb8129b597d7dbd00d : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2a1439720049b7a5cccddc2cb40b16 : SUCCESS in 7m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4af975ea064142cbb6c0b0a15f40bf37 : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":8},{"id":"67f67aa9ad89865bfbd7e3cdec5e4eed47cd090a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 15:25:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"06b97ad2f6b72689be04281e835b53976decb3c7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-06 01:55:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1fde8dc290a537a231d2c2cd78e170d803baf704","revisions":{"facffcafe092d6e48c0eb750c2bc1a61f4d9d3c9":{"kind":"REWORK","_number":1,"created":"2023-03-28 02:16:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/878731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/878731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/878731/1"}}},"commit":{"parents":[{"commit":"858ce8ee6eb3d2c50bfeba32aae0b252ac8ffe04","subject":"Switch back Puppet OpenStack to cycle-with-intermediary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/858ce8ee6eb3d2c50bfeba32aae0b252ac8ffe04"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:11:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:15:32.000000000","tz":540},"subject":"Puppet OpenStack 2023.1 release","message":"Puppet OpenStack 2023.1 release\n\nChange-Id: I536d9cdd8933e3cc7aece42e129bd9d42b4bc0f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/facffcafe092d6e48c0eb750c2bc1a61f4d9d3c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/facffcafe092d6e48c0eb750c2bc1a61f4d9d3c9"}]},"branch":"refs/heads/master"},"35c6bd3c03b3ea07eebd47bba5d4d729d8fddd94":{"kind":"REWORK","_number":2,"created":"2023-03-28 02:17:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/878731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/878731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/878731/2"}}},"commit":{"parents":[{"commit":"858ce8ee6eb3d2c50bfeba32aae0b252ac8ffe04","subject":"Switch back Puppet OpenStack to cycle-with-intermediary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/858ce8ee6eb3d2c50bfeba32aae0b252ac8ffe04"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:11:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:17:34.000000000","tz":540},"subject":"Puppet OpenStack 2023.1 release","message":"Puppet OpenStack 2023.1 release\n\nChange-Id: I536d9cdd8933e3cc7aece42e129bd9d42b4bc0f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/35c6bd3c03b3ea07eebd47bba5d4d729d8fddd94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/35c6bd3c03b3ea07eebd47bba5d4d729d8fddd94"}]},"branch":"refs/heads/master"},"282c88812cb68105772ff340e93eca4da173fb76":{"kind":"REWORK","_number":3,"created":"2023-03-28 02:18:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/878731/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/878731/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/878731/3"}}},"commit":{"parents":[{"commit":"858ce8ee6eb3d2c50bfeba32aae0b252ac8ffe04","subject":"Switch back Puppet OpenStack to cycle-with-intermediary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/858ce8ee6eb3d2c50bfeba32aae0b252ac8ffe04"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:11:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:18:50.000000000","tz":540},"subject":"Puppet OpenStack 2023.1 release","message":"Puppet OpenStack 2023.1 release\n\nChange-Id: I536d9cdd8933e3cc7aece42e129bd9d42b4bc0f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/282c88812cb68105772ff340e93eca4da173fb76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/282c88812cb68105772ff340e93eca4da173fb76"}]},"branch":"refs/heads/master"},"287ddd0e2a8ff089118f3c482667f0f33293632b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-03-28 14:01:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/878731/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/878731/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/878731/4"}}},"commit":{"parents":[{"commit":"964990c35cda830af83d11097c273c0f3c54a2da","subject":"Switch back Puppet OpenStack to cycle-with-intermediary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/964990c35cda830af83d11097c273c0f3c54a2da"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:11:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 14:01:25.000000000","tz":540},"subject":"Puppet OpenStack 2023.1 release","message":"Puppet OpenStack 2023.1 release\n\nChange-Id: I536d9cdd8933e3cc7aece42e129bd9d42b4bc0f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/287ddd0e2a8ff089118f3c482667f0f33293632b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/287ddd0e2a8ff089118f3c482667f0f33293632b"}]},"branch":"refs/heads/master"},"3d5066af719dec41d9d75efa1de8fa394a0e0515":{"kind":"REWORK","_number":5,"created":"2023-03-29 06:16:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/878731/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/878731/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/878731/5"}}},"commit":{"parents":[{"commit":"964990c35cda830af83d11097c273c0f3c54a2da","subject":"Switch back Puppet OpenStack to cycle-with-intermediary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/964990c35cda830af83d11097c273c0f3c54a2da"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:11:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-29 06:16:45.000000000","tz":540},"subject":"Puppet OpenStack 2023.1 release","message":"Puppet OpenStack 2023.1 release\n\nChange-Id: I536d9cdd8933e3cc7aece42e129bd9d42b4bc0f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3d5066af719dec41d9d75efa1de8fa394a0e0515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3d5066af719dec41d9d75efa1de8fa394a0e0515"}]},"branch":"refs/heads/master"},"b4777aa6bce2a1820cf6fe7022a0d612f35fa5b4":{"kind":"REWORK","_number":6,"created":"2023-03-29 06:43:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/878731/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/878731/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/878731/6"}}},"commit":{"parents":[{"commit":"964990c35cda830af83d11097c273c0f3c54a2da","subject":"Switch back Puppet OpenStack to cycle-with-intermediary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/964990c35cda830af83d11097c273c0f3c54a2da"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:11:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-29 06:43:11.000000000","tz":540},"subject":"Puppet OpenStack 2023.1 release","message":"Puppet OpenStack 2023.1 release\n\nChange-Id: I536d9cdd8933e3cc7aece42e129bd9d42b4bc0f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4777aa6bce2a1820cf6fe7022a0d612f35fa5b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4777aa6bce2a1820cf6fe7022a0d612f35fa5b4"}]},"branch":"refs/heads/master"},"a472bd8f3b2ff1c55b3bcf3bd91b77bccb02fd4f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-04-03 00:19:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/878731/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/878731/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/878731/7"}}},"commit":{"parents":[{"commit":"00b2b84c48a584839e461abcc3fba79873540853","subject":"Add workaround for announce mail script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00b2b84c48a584839e461abcc3fba79873540853"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:11:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-03 00:19:29.000000000","tz":0},"subject":"Puppet OpenStack 2023.1 release","message":"Puppet OpenStack 2023.1 release\n\nChange-Id: I536d9cdd8933e3cc7aece42e129bd9d42b4bc0f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a472bd8f3b2ff1c55b3bcf3bd91b77bccb02fd4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a472bd8f3b2ff1c55b3bcf3bd91b77bccb02fd4f"}]},"branch":"refs/heads/master","description":"Rebase"},"1fde8dc290a537a231d2c2cd78e170d803baf704":{"kind":"REWORK","_number":8,"created":"2023-04-03 01:54:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/878731/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/878731/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/878731/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/878731/8"}}},"commit":{"parents":[{"commit":"00b2b84c48a584839e461abcc3fba79873540853","subject":"Add workaround for announce mail script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00b2b84c48a584839e461abcc3fba79873540853"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-28 02:11:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-03 01:54:39.000000000","tz":0},"subject":"Puppet OpenStack 2023.1 release","message":"Puppet OpenStack 2023.1 release\n\nChange-Id: I536d9cdd8933e3cc7aece42e129bd9d42b4bc0f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1fde8dc290a537a231d2c2cd78e170d803baf704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1fde8dc290a537a231d2c2cd78e170d803baf704"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:PTL-Approved\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
