)]}'
{"id":"openstack%2Freleases~879273","triplet_id":"openstack%2Freleases~master~I52714ae92458efc0628091762cb77cf3ddb058f9","project":"openstack/releases","branch":"master","topic":"fix-announce","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2023-06-01 12:02:09.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-04-11 10:28:22.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-06-01 14:46:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I52714ae92458efc0628091762cb77cf3ddb058f9","subject":"Fix tools/announce by using importlib.metadata instead of setuptools","status":"MERGED","created":"2023-04-03 08:42:01.000000000","updated":"2023-06-01 14:46:17.000000000","submitted":"2023-06-01 14:46:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"879273-fix-announce","meta_rev_id":"3237d3bc6b53fdf27154df081cd0d41839de82bb","_number":879273,"virtual_id_number":879273,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-01 14:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-06-01 12:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-06-01 13:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-06-01 13:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-06-01 14:35:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 08:57:21.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-04-03 09:02:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 13:39:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-04-06 07:43:47.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-06-01 13:00:24.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"243538ef5297e7ba7e4690a1274ced1d0d30d643","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-03 08:42:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0e938a1a8c9d66f912d25d78fab85012663c206","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-03 08:43:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bdf13c84fea27fa98ee903efa3819b8ae0d289eb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-03 08:57:21.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0ceac99bf1bd64de6c8589eda7cb54f2da0324d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 09:02:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df602172269f43efb2ea1307ca988093\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eeadbb98883455ba83ffa4aced4d43d : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/692aee67c30f43a6b0b13cbf10579572 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4753d98deb8942dd9a9eb0c1f7921d3b : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b687620224a344a6b0712bfe44efe4f5 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2dd350da33454585b25b54a4a137e4 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2831c38c9c5045889cf6a64089aedc12 : SUCCESS in 4m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/98e5af622cde4357b051f2bedbbde0e7 : SUCCESS in 3m 14s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/65a5accd69ee48fb9b970a031ff2073b : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"970a2104b31db609eb783774f3e4a3db09e00872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-03 12:35:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8fb5746411deba888ed6a14b2917b1669b6b1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 12:50:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9cfacbdd77943e3b7bc8f4ca0017bf9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a984d1292f344212b504e9023389e74d : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e756ba3bd35b44c2a79f2488a9f2e390 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3f16ca85f6d4b7b89c94b1370144c23 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45f5ac2304194c06b1b96b8a2f42779b : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f697a37d20c24b63b045975f1e09430b : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d52fd39af3c406e93b3fb770de32550 : SUCCESS in 4m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/40f0eb8104b447a2ac18bbbf96621f9e : SUCCESS in 3m 58s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f2abcc946c5743029123b7fafd05f7a3 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"d26558867c7f3014ee6b62b34267b4d862b04951","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-04 13:39:49.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3de0473f199936f4cf9baf271ad1a4bb79c08e89","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-05 07:02:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e2435a73d4dee056da563b30780caefbbae45fe","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-06 07:43:47.000000000","message":"Patch Set 2: Code-Review+2 PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"1db58bf62d2ce8d283e749f622ae1579b4f49dfe","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-06 10:13:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d7ebb7cd75c1183b551f91f3cafaba4f2289749","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-06 12:57:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"efa9751a09456c1baebf08d9af9bae77f90e84fc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-06 13:45:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b20a558d6ddf2fb34ff48aded2f8c7010a49734","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-07 13:53:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3af9eb41fe36b6fe7549f9fcc6a2f8d4cb578a88","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-11 10:28:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6687726ada1b353516783d28792cf96560962ef","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-11 13:16:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50d4e6d1975cc5e880b4b86d746be1c12e5a13ad","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-12 09:39:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e249830c5704317dca4da03752cadad60a868f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-22 08:22:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7a9c47f0c75f2403619bf6c8fef46530f5f2c7b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 08:32:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c72fec1b4194642aed796ca9c4a6613\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa1ae6cb77c54092a2bf86b80486b4f7 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04dad4ae7098459996ee82c7d166b3ca : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8cc55313bf64b1ba4d3a194eeccdfbb : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df270af802aa4487b50132293c15f9df : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbbe320182aa481ba4fd8e28e36f9820 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/843c9a60e7b54bb1bcff3ae588cd005b : SUCCESS in 4m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f6eae927831b44059ccba301e3c422b8 : SUCCESS in 3m 01s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/62c9ea93757848468f555a92049b1be0 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"37acb04a75441297f804934385576eb23cf1be22","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-06-01 12:02:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5ff11d6da62d64ed266097d42fe847204b0cb0fa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-06-01 13:00:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"08e60d292dc79f67a79a5e98e6ec96caf448b4de","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-06-01 14:35:31.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"5a636af08a82ffbd72c9e77f0860e40789a3dfab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 14:35:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"853936272a470d037ef0a9de28c66737e5db6b32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 14:46:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6666300ab18b43c8b91da0be0b12949c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42a3dd91befb41e6852150d17481f2d1 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5983c876c4cb463096f770b967a4d614 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60656bf6d12943ecacbfeabf655f4646 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18396a8de9d54dc4b413970c505bb0a1 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96e65f7130645fa96032a846d22396e : SUCCESS in 7m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9375ed92a0c94daa97a3e8baa09823c9 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"3237d3bc6b53fdf27154df081cd0d41839de82bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 14:46:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b46589cc395062e328a204e33ebd0c8b49bac56","revisions":{"559d3f6291e4c02b561644bdd4295b4d6dda017e":{"kind":"REWORK","_number":1,"created":"2023-04-03 08:42:01.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/73/879273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/879273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/879273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/879273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/879273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/879273/1"}}},"commit":{"parents":[{"commit":"00deff90c0c0fda03101853753d314b07d5229da","subject":"Merge \"Add workaround for announce mail script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00deff90c0c0fda03101853753d314b07d5229da"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 08:37:17.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 08:41:52.000000000","tz":120},"subject":"Fix tools/announce by using importlib.metadata instead of setuptools","message":"Fix tools/announce by using importlib.metadata instead of setuptools\n\nWith release of new setuptools running setup.py started\nreturining extra line. Eventually, running setup.py as a script is not\nconsidered as good practice and we should not rely on it\u0027s output much.\nInstead, we can leverage importlib.metadata as it should be more\nreliable in long run.\n\nAlso it allow us to remove our previous workaround:\nhttps://review.opendev.org/c/openstack/releases/+/879065\n\nChange-Id: I52714ae92458efc0628091762cb77cf3ddb058f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/559d3f6291e4c02b561644bdd4295b4d6dda017e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/559d3f6291e4c02b561644bdd4295b4d6dda017e"}]},"branch":"refs/heads/master"},"fd6d5b78d68d30f87c25057b5182074c5cd6dc78":{"kind":"REWORK","_number":2,"created":"2023-04-03 12:35:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/73/879273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/879273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/879273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/879273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/879273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/879273/2"}}},"commit":{"parents":[{"commit":"00deff90c0c0fda03101853753d314b07d5229da","subject":"Merge \"Add workaround for announce mail script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00deff90c0c0fda03101853753d314b07d5229da"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 08:37:17.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 12:35:31.000000000","tz":120},"subject":"Fix tools/announce by using importlib.metadata instead of setuptools","message":"Fix tools/announce by using importlib.metadata instead of setuptools\n\nWith release of new setuptools running setup.py started\nreturining extra line. Eventually, running setup.py as a script is not\nconsidered as good practice and we should not rely on it\u0027s output much.\nInstead, we can leverage importlib.metadata as it should be more\nreliable in long run.\n\nAlso it allow us to remove our previous workaround:\nhttps://review.opendev.org/c/openstack/releases/+/879065\n\nChange-Id: I52714ae92458efc0628091762cb77cf3ddb058f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fd6d5b78d68d30f87c25057b5182074c5cd6dc78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fd6d5b78d68d30f87c25057b5182074c5cd6dc78"}]},"branch":"refs/heads/master"},"2b46589cc395062e328a204e33ebd0c8b49bac56":{"kind":"REWORK","_number":3,"created":"2023-05-22 08:22:22.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/73/879273/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/879273/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/879273/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/879273/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/879273/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/879273/3"}}},"commit":{"parents":[{"commit":"00deff90c0c0fda03101853753d314b07d5229da","subject":"Merge \"Add workaround for announce mail script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00deff90c0c0fda03101853753d314b07d5229da"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 08:37:17.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-05-22 08:22:09.000000000","tz":120},"subject":"Fix tools/announce by using importlib.metadata instead of setuptools","message":"Fix tools/announce by using importlib.metadata instead of setuptools\n\nWith release of new setuptools running setup.py started\nreturining extra line. Eventually, running setup.py as a script is not\nconsidered as good practice and we should not rely on it\u0027s output much.\nInstead, we can leverage importlib.metadata as it should be more\nreliable in long run.\n\nAlso it allow us to remove our previous workaround:\nhttps://review.opendev.org/c/openstack/releases/+/879065\n\nChange-Id: I52714ae92458efc0628091762cb77cf3ddb058f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b46589cc395062e328a204e33ebd0c8b49bac56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b46589cc395062e328a204e33ebd0c8b49bac56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
