)]}'
{"id":"openstack%2Freleases~880631","triplet_id":"openstack%2Freleases~master~I93555dc090ceb8a024f26602ed339814f7dbd0b0","project":"openstack/releases","branch":"master","topic":"openstacksdk_1.1.0","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-27 15:00:50.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-04-27 15:00:50.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-04-18 10:50:21.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-04-20 09:11:14.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I93555dc090ceb8a024f26602ed339814f7dbd0b0","subject":"Release OpenStackSDK 1.1.0","status":"MERGED","created":"2023-04-17 13:39:55.000000000","updated":"2023-04-27 15:00:50.000000000","submitted":"2023-04-27 15:00:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880631-openstacksdk_1.1.0","meta_rev_id":"b9c50a74f912e6b3662cfb3bb8f16db6209a78d6","_number":880631,"virtual_id_number":880631,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-27 15:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-04-27 14:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-04-20 12:13:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-04-20 09:10:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-04-20 09:11:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-04-27 14:46:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-04-20 09:12:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 13:40:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-17 13:56:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 14:20:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2023-04-17 17:01:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-04-18 08:04:10.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-04-20 09:10:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-04-27 14:46:46.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"daeaf12c40b88a11253778e549e0cce36fe29a37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-17 13:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cda96d7e440930c699c6b3d102ad2435f4469b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:56:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1084b601bb74ba986f8e8b031c11e6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c43466c5155a4f0ebe1a447ee10bd67e : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/caacbd20515b41b5b20e906c9807f96c : FAILURE in 6m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/82ea1621ec0048f596cdd3c2d7237d9b : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e478f4a4ff2843080b198b74f4f1658db682ae6f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-17 16:05:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b09e2e1180d5de8635da1c9d14b74776afe6d8bc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-17 16:05:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a0a308aee613fc21a624f7f789f31778637ba58","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-17 17:01:18.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fb6918087452490ad0c6f400e299112ff86d6e2b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-18 08:04:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e00512bfa31dd7f21d19ce4f08607a0cc345170","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-18 09:32:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5da6c311d817c5e787768580c7dc369cd46e4d54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-18 09:33:03.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0e3ded8ca2aa4c18e3f413c48ef993424942c70c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 09:41:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16394b03fb2e4c4db0be4b8ad7a760ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3e5b7ec3304943b57c06915cd33920 : SUCCESS in 6m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c50c7cd424e4d41aaa9ddb3bf70ace2 : SUCCESS in 6m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1f368761e8c34d2b98eab18fc1a015af : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":2},{"id":"0ea8d59d2f1f70c66028d231cf066370b0341e12","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-18 10:50:21.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c9b4296f29658eb689af94bf8eaf1e850ef6788e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-20 09:10:17.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"711dc436a13dfef57e8229fe0237e6b22a033f0f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-04-20 09:11:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"650d487475461b82b9f0f1b27f5e3420ec3f1151","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 09:12:16.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"1a33bdcd046718367b14a10cdc1300bebd3709ba","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-20 12:13:33.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0adeb705ae6c1375de554cacce5192e8189787b9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-24 15:35:21.000000000","message":"Topic openstacksdk_1.0.2 removed","accounts_in_message":[],"_revision_number":2},{"id":"657c83b2b0508e51f0afd57c32d938494774d7be","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-24 15:35:42.000000000","message":"Topic set to openstacksdk_1.1.0","accounts_in_message":[],"_revision_number":2},{"id":"e9c8ff27b12caedcf55c5f5a914d85c889aecbdd","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-27 14:46:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8c920b3245c523ff705211fd12491227dc42fd13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 14:47:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8e34061f2edf2fa20d99b61a875d91d4542d8d3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 15:00:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c1ae99bbf724f6b8d69330aedbbcb64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b304271a88043cc8da491710f37e9c5 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd28ad40f6340afbb7b025e6e11555d : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03bbff1cd478443bb443461ba40d0dbc : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b9c50a74f912e6b3662cfb3bb8f16db6209a78d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 15:00:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b4f444958da619a29bd688052b71e5f7dd3368b5","revisions":{"17c8f4cc40ab50f410561bb526b6e1ddfccf85b1":{"kind":"REWORK","_number":1,"created":"2023-04-17 13:39:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/880631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/880631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/880631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/880631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/880631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/880631/1"}}},"commit":{"parents":[{"commit":"a327abe4eb12cb19bc31ccb3c30a6349d5d3427e","subject":"Merge \"Retire patrol\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a327abe4eb12cb19bc31ccb3c30a6349d5d3427e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-04-15 00:04:28.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-04-15 00:04:28.000000000","tz":120},"subject":"Release OpenStackSDK 1.0.2","message":"Release OpenStackSDK 1.0.2\n\n$ git log --oneline --no-merges 1.0.1..26797197\nc3e77fc6 Add resize/extend share actions.\necde4028 Add export location resource to shared file system\nac194f98 Microversion 2.91: Support specifying destination host to unshelve\na9614648 Adds Support for ``glance cache-clear``\n5b3ad879 Removing region Lon1. Updating block_storage_api_version. Adding image_format.\n1921a195 compute: Add Server.restore, restore_server proxy method\n53226205 ssh key change\n6b0206a5 Add Cleura acceptance tests\n7bbef2f0 Change python3 jobs template to latest version\na2f3b54f Drop legacy job that is always failing\nb0d98dbb Rework zuul config\n50711b46 Prepare acceptance tests for real clouds\n1ca0f0c3 Add share instances to shared file systems\n54b25722 Use custom warnings, not logging.warning\n27e66984 Adding support for glance cache-queue Command\ndff6625f config: Load additional options for v3multifactor\n76724972 config: Split \u0027OS_AUTH_METHODS\u0027\n61422fd8 add a new vendor profile for the Swiss Open Telekom Cloud\n79e8c83d Add share snapshot instance resource\nb06c27b9 Adds SDK support for ``glance cache-delete``\n1b241d74 Add BGPVPN to SDK\n43ab59d8 Implement acceptance test job\n9f073752 Fix Accept header for deleting Octavia load balancers\nc2b8612a Add share network resource to shared file system\n7c7092e7 Update master for stable/2023.1\n74f8869f Remove \"feature/r1\" override from manila job\n8086da0d Include \"security_groups\" to \"Port\" query parameters\nc0e8027e Add support for glance cache\n8588a4c7 Remove usage of deprecated `sre_constants` module\n4ce2e960 fix docstrings refering to volume attachments instead of server migrations\n163b6944 Revert \"Revert \"Add \"security_group_ids\" to Port\u0027s query parameters\"\"\n4ad5c174 cloud: create_server: fix scheduler_hints/group\n400fbfbe support None as expected status in wait_for_status\n\nChange-Id: I93555dc090ceb8a024f26602ed339814f7dbd0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/17c8f4cc40ab50f410561bb526b6e1ddfccf85b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/17c8f4cc40ab50f410561bb526b6e1ddfccf85b1"}]},"branch":"refs/heads/master"},"b4f444958da619a29bd688052b71e5f7dd3368b5":{"kind":"REWORK","_number":2,"created":"2023-04-18 09:33:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/880631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/880631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/880631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/880631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/880631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/880631/2"}}},"commit":{"parents":[{"commit":"a327abe4eb12cb19bc31ccb3c30a6349d5d3427e","subject":"Merge \"Retire patrol\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a327abe4eb12cb19bc31ccb3c30a6349d5d3427e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-04-15 00:04:28.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-04-18 09:33:03.000000000","tz":0},"subject":"Release OpenStackSDK 1.1.0","message":"Release OpenStackSDK 1.1.0\n\n$ git log --oneline --no-merges 1.0.1..26797197\nc3e77fc6 Add resize/extend share actions.\necde4028 Add export location resource to shared file system\nac194f98 Microversion 2.91: Support specifying destination host to unshelve\na9614648 Adds Support for ``glance cache-clear``\n5b3ad879 Removing region Lon1. Updating block_storage_api_version. Adding image_format.\n1921a195 compute: Add Server.restore, restore_server proxy method\n53226205 ssh key change\n6b0206a5 Add Cleura acceptance tests\n7bbef2f0 Change python3 jobs template to latest version\na2f3b54f Drop legacy job that is always failing\nb0d98dbb Rework zuul config\n50711b46 Prepare acceptance tests for real clouds\n1ca0f0c3 Add share instances to shared file systems\n54b25722 Use custom warnings, not logging.warning\n27e66984 Adding support for glance cache-queue Command\ndff6625f config: Load additional options for v3multifactor\n76724972 config: Split \u0027OS_AUTH_METHODS\u0027\n61422fd8 add a new vendor profile for the Swiss Open Telekom Cloud\n79e8c83d Add share snapshot instance resource\nb06c27b9 Adds SDK support for ``glance cache-delete``\n1b241d74 Add BGPVPN to SDK\n43ab59d8 Implement acceptance test job\n9f073752 Fix Accept header for deleting Octavia load balancers\nc2b8612a Add share network resource to shared file system\n7c7092e7 Update master for stable/2023.1\n74f8869f Remove \"feature/r1\" override from manila job\n8086da0d Include \"security_groups\" to \"Port\" query parameters\nc0e8027e Add support for glance cache\n8588a4c7 Remove usage of deprecated `sre_constants` module\n4ce2e960 fix docstrings refering to volume attachments instead of server migrations\n163b6944 Revert \"Revert \"Add \"security_group_ids\" to Port\u0027s query parameters\"\"\n4ad5c174 cloud: create_server: fix scheduler_hints/group\n400fbfbe support None as expected status in wait_for_status\n\nChange-Id: I93555dc090ceb8a024f26602ed339814f7dbd0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4f444958da619a29bd688052b71e5f7dd3368b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4f444958da619a29bd688052b71e5f7dd3368b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
