)]}'
{"id":"openstack%2Freleases~882819","triplet_id":"openstack%2Freleases~master~Iea7bef0e3e3915bea6859dabccdd032bba6717ae","project":"openstack/releases","branch":"master","topic":"neutron_23.0.0.0b1","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-05-11 08:48:34.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-05-10 12:01:30.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-05-11 08:48:34.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-05-11 08:48:34.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-10 12:43:32.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-05-11 08:48:34.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2023-05-11 08:48:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea7bef0e3e3915bea6859dabccdd032bba6717ae","subject":"Release Neutron 23.0.0.0b1 (Neutron Bobcat beta 1)","status":"MERGED","created":"2023-05-10 11:21:06.000000000","updated":"2023-05-11 08:48:34.000000000","submitted":"2023-05-11 08:48:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882819-neutron_23.0.0.0b1","meta_rev_id":"2b74221c3adf76eddcebea9f4f6eae4434ff53cb","_number":882819,"virtual_id_number":882819,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-11 08:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-05-10 11:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-05-10 14:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-05-11 08:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-05-10 12:01:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-05-10 12:43:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-05-11 08:42:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-05-10 11:23:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-10 11:21:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-05-10 11:21:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2023-05-10 11:21:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-05-10 11:21:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-05-10 11:21:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-05-10 11:21:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-05-10 11:21:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 14:52:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-05-11 08:42:09.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"12cb45b8ba9f1c1f4d6bd0cbc65fb45d458401b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-10 11:21:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40f8bdfad52c1d73d674391872ca33a94b0b5d1a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-10 11:21:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fc27adf92e446c6ae8050e81fab214ccec4ea99f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 11:21:55.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"eb8557c9836e17bf79e17ddd8eb012abed5c9f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-10 11:22:37.000000000","message":"Patch Set 2: Commit message was updated.\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f816551736436477ff7dc213683f2c3d1633d2f4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-10 11:22:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ae6122bc5ead4781fd6eefca3316ab91d7833ec6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 11:23:22.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9f79a0cd1ce95bae9e9917d0321e3f83e3f3eef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 11:30:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/909a632c8b0a44bb9d57dee64a4f5668\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5148816748a545989bfe8086582cc959 : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc19cae2021b4cb6b33ab332aece9cb4 : SUCCESS in 5m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1d61f0864e12434a88d07fd715114cb5 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"181132db7a66efa8941b74963ffd213d088399c7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-05-10 12:01:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c663f5df55568512044a5d1e8aeea55fd61d8820","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-10 12:43:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bae5286a77bca82ccd756e88e13e8e6fdf2156ff","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-05-10 14:52:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"373018c54ab18ff1bd2713f15e2da2da975ee7f9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-05-11 08:42:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"64190b444a40d76e66275fb757bb2eef9bfd7808","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 08:42:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"397eea39300420ce51fb9cd0a4ff237ee78dd102","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 08:48:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92e57c2e5acc4f7d8b8218cedfcbaeaa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/011a9228cc304a4e919218ab722cbfa6 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f35c82a03e0c41d7befca5fd10270825 : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa12a488396545f28b20cf4b1c30f4a3 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"2b74221c3adf76eddcebea9f4f6eae4434ff53cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 08:48:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c1e7fefc44deb39a98d0e5c820c94dea23d88a4","revisions":{"0693572151c1343be30fbdc0d1b2f5df2aa537e5":{"kind":"REWORK","_number":1,"created":"2023-05-10 11:21:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/19/882819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/882819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/882819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/882819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/882819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/882819/1"}}},"commit":{"parents":[{"commit":"8e7f1505afa2cf5e0be35a0e460d3367173233bf","subject":"Merge \"[octavia] Transition Xena to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e7f1505afa2cf5e0be35a0e460d3367173233bf"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-09 21:28:03.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-09 21:28:03.000000000","tz":120},"subject":"Release Neutron 23.0.0.0b1 (Neutro Bobcat beta 1)","message":"Release Neutron 23.0.0.0b1 (Neutro Bobcat beta 1)\n\nThis release contains the change [1] that enables by default\nthe new sRBAC policies.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/879827\n\n$ git log --oneline --no-merges 21.0.0.0rc1..9319ba00a\nc3602ac19b [OVN] Update ovn meter when neutron server reloads\nc0af5b3b5e Reduce lock contention on subnets\n256297fc7f rbacs: clean-up to use defined constants ACCESS_*\n6eb7006801 Drop retries in tests for TimeoutException\n30c0e5699e Fix doc links for networking option 2\nb1cc242fad Add a method to retrieve router gateway ports\n7a59cf0eb9 Correct planned removal for maintenance function\n2be53b1719 [functional] Fix db_set use in test_cascading_del_in_txn\na9963e90d9 ``_get_ovn_version`` returns a 3 element tuple\n4ad979a534 Mark \"test_port_creation_and_deletion\" as unstable\n9ac59e4b4a Avoid retrieving ports if network list is empty\n4bac350f68 Remove \"neutron-ovn-tempest-ovs-release-ubuntu-old\" job\nb19b55909d Don\u0027t set and remove immediately DEAD VLAN tag in tests\n88ce859b56 Change API to validate network MTU minimums\n9f22dc1d3a Doc: Add FWaaS v2 install details\n853dc2570d Add py39 jobs to tox override template\n39d252da7c Deprecated support for Windows OS\n82029c2c51 Use a writer context for the online alembic migrations\n2cafe0a9c4 Revert \"Move to python3.9 as minimal python version\"\nf93c9be1c1 Move to python3.9 as minimal python version\n4e27e27ae2 Replace context decorators with context managers\ndd184c5c10 Fix Loki tempest jobs\nb31453af47 [OVN] Admin procedure for duplicated or deleted OVN agents\n872b53f618 Fix functional tests modules which are using PluginFixture\n670cc383e0 [S-RBAC] Switch to new policies by default\nd757c530bc Update QoS documentation\na84567b8d6 Remove the ``OVNSqlFixture`` class workaround\n05ba4257de Remove creation of DHCP port in the OVN metadata unit test\n43829301f3 Handle \"no such process\" during keepalived process cleanup\n34ea8988d6 [OVN] Add update event to ``OVSInterfaceEvent`` class\na5e26408d2 Fix dns_integration and ml2 plugin unit tests modules\n5db17654b9 Fix servicetype unit tests module\nf92d6fa72a Fix segments unit tests module\n6b5acb5835 [S-RBAC] Get availability zone API available for READER role\n340fd3cfe7 Run master jobs only on master branch\n706a0e0268 Fix parent of neutron-ovn-tempest-with-uwsgi-loki\n28961c8b76 Fix network_segment_range unit tests module\n73ac4510c7 [grenade] Collect ovn services logs\n2364327bd1 Run periodic jobs in experimental queue too\n9d51633013 Fix intermittent failures in finding metada port in SB DB\n267efd2984 OVN: Always try and create a metadata port on subnets\nc7ef824941 Do not check the context object in ``TestMeteringPlugin``\n5510cdab92 [ovn] OVNClient._get_router_ports: Drop unused parameter\ne5d4499672 [ovn] Drop use of LR OVN_GW_NETWORK_EXT_ID_KEY\nd67d1c2736 [ovn] Drop use of OVN_GW_PORT_EXT_ID_KEY\n69f30c92ef [sqlalchemy-20] Add reader context to\n``get_ports_on_host_by_subnet``\n7dfbdf65a7 Add support for localnet_learn_fdb OVN option\ne374b82d8f [CI][fullstack/functional] Report slowest tests\n8eecccfeae [S-RBAC] Allow network owners to get ports from that network\n95c19c8868 Checkout \"sqlalchemy/alembic\" main branch in\nsqlalchmey-master jobs\n3de8ebebd8 Pin OVS_BRANCH to working commit\na06fe7cfd9 Fix NoSuchOptError error in Ipam unit tests\n5a17f2b24a Pass physical bridge informations to OVS agent extension API\n7828acaf4f [ovn] Add end to end test for QosExtension\n2aee961ab6 Suppress IPv6 metadata DAD failure and delete address\nade2a9f893 Mark \"test_multiple_agents_for_network\" as unstable\nfa172ab7dc rbacs: fix typo, s/cxt/ctx\ne6de524555 rbacs: filter out model that are already owned by context\n7073410be3 Bump skip-level lower version to stable/zed\n775d5de9f1 Add debug information to\n``MacvtapAgentTestCase.test_get_all_devices``\n0ec04dd638 Ensure redirect-type\u003dbridged not used for geneve networks\nd4654e3011 Filter out unsatisfied routers in SQL\n33c4a2d97e Update url and package name\n3cc28a004a Add plugin.spec to irrelevant-files\nb777aa57b2 Update the quota guide examples\ncfe38a0014 Change name of ``_TestIsSessionActive`` test case.\n27b3eacd3d Revert \"Ensure vlan network traffic is not centralized\"\ne4da60740b [sqlalchemy-20] Do not use strings for aatribute names in\nloader options\n0a69dd5e3d [sqlalchemy-20] query.join should define one table/column per\ncall\n4b8e484e1d Increase port name size and type to internal\n2ccf1e1e90 [OVN] OVN agent should register \"Chassis_Private\" by default\nf42f1cfa69 [sqlalchemy-20] Provide SQL \"case\" expression correct input\nparemeters\n831ac3152d Fix a number of configuration typos\n5b4ed5b117 Fix concurrent port binding activate\n1646e5b28d Add neutron-lib project to the SQLAlchemy master branch CI\njobs\n0220236c63 Cleanup before executing \"test_get_all_devices\"\n6e1dbe9781 Add oslo.db project to the SQLAlchemy master branch CI jobs\n97aa84b69a Open the 2023.2 (Bobcat) DB branch\nc97dcfd03f doc: state that O flag can be 0 in dhcpv6-stateful\n6358495720 Delete sg rule which remote is the deleted sg\n04d3f889ef Fix metadata agent intermittent test failures\n9704dca84e [OVN] Explicitly define the fixed IPs for the metadata port\ndea48cfc0a Only create a frozen  Row on matching events\n711fbb9820 Imported Translations from Zanata\n8ea2a9c128 Fix typo in unit test\n442b437a81 Change external process manager tests to clean temp files\nefab60c0bf Try to optimize Mysql server mem usage on some CI jobs\n745497a112 [OVN] Remove \"update_port_qos_with_external_ids_reference\"\n5c98d9e8d1 [OVS] Parse the \"permitted_ethertypes\" at the FW\ninitialization\nbffa642b35 [OVN] Method ``get_port_qos`` should always return 2 values\n008277b8c1 [OVS] Allow custom ethertype traffic in the ingress table\nb9567033fc [OVN] Use the BW values retrieved from ``get_port_qos``\n08eb8e2498 Use neutron-lib policy rules\nc8ccf2ffbb [OVN] Change oslo config options entry point for the OVN\nagent\ncf96bd8bdf ovs: fix regression when vlan mapping is not already\nregistered\nd44f164f4d ovn_idl_impl: fix a logic bug in get_sg_port_groups\nd9358b67bd functional: set dns_domain config option after its\nregistration\n8946684fb2 Remove duplicate rows in MySQL query output\n5d2086c698 Add 2023.1 release name in routed networks doc\nb6bc4c8a66 Add Lajos Katona to Client and Doc areas as lieutenant\nd4a85833a7 [sqlalchemy-20] The Session.begin.subtransactions flag is\ndeprecated\n0a214b0437 Imported Translations from Zanata\n999116126e Add full support for OVN NB \"Gateway_Chassis\" table\n4254ccd1bc Update master for stable/2023.1\n8e3bddbf8b Ensure vlan network traffic is not centralized\n39b65575cd Change the release tag to use the release identification\n44ac03de14 Fix policy unit test deprecation warnings\nebcde41fc8 [OVN] Add ``get_gateway_chassis_az_hints`` method to OVN API\n41846a6e8b [sqlalchemy-20] The Connection.connect() method is considered\nlegacy\n2af5fd889b Add sleep before checking if ovs port is in the namespace\n12093015de Add Jens Harbott as Lieutenants in Infra area\n19482a049b Add Slawomir Kaplonski as Lieutenant in API, Client and\nTesting areas\n51005388b8 Make retrieval of port mac column safe\n6cac935824 Update lieutenants for ovn-octavia-provider\n948c9e02e3 [S-RBAC] Add release note about full support for new policies\n289ae97c1c Add missing osprofiler options\nacb809eea4 [OVN] Ensure traffic for provider vlan networks is not\ntunneled\n7f46ea9b6d Remove unused rbac rule\n75e8360224 Reintroduce agent bridge resync test\nc2a02d3c69 Remove two duplicated unit tests\ndbfc18d1fa Apply Ironic\u0027s server-ip-address as TFTP next-server\nbdf0c8bdad Move register_common_config_options call to DietTestCase\ne7c92d92e4 [sqlalchemy-20] Use the .begin() method of Engine\n827fbd01c3 Normalise format of OVN agent heartbeat timestamp\ne68e4162ce Prevent router_ha_interface port from being removed via API\naa40aef70f Mark fullstack TestMultiSegs.test_multi_segs_network as\nunstable\n169ef05589 Change neutron-ovs-tempest-dvr-ha-multinode-full job\u0027s config\nb00ae45140 ml2 plugin: use const from neutron-lib\nf991fc2fbe [ovn][ipv6] Skip test_update_delete_extra_route\n1e41558851 Reduce number of routers created in\n``test_gateway_chassis_rebalance``\n51d449babf Move tox-py311 job override into job-templates.yaml\nf5dc708e1a Check port.tag is not DEAD_VLAN_TAG in\n``DHCPAgentOVSTestFramework``\nc61ce3447e Format correctly (dialect\u003dmac_unix_expanded) the MAC\naddresses\n502c0253de [ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class\ne9ce025a69 [dhcp agent] Fix clean_devices\nd052da85cf Mark ``test_port_creation_and_deletion`` as unstable\n5e15c4a4d4 Allow other kwargs in ``DhcpBase.disable``\n5dac5f8c6c [ovn-ipv6] Temporary exclude test_hotplug_nic\n67a232fe71 Remove tripleo related jobs from neutron queues\n595c387efb [UT] Mock the retry decorator when creating duplicated\nNDPProxy\nf9880c2013 Increase tox-py311 timeout to 3600 seconds\n24152f2019 Add debugging exception messages in\n``test_ovs_and_ovs_events``\nfce516e346 [OVN] Bump the port revision number in trunk driver\nd83e243c6d [OVN] Add missing LSP device_owner info in trunk driver\n76f3b75ad6 [OVN] Skip ports with VNIC type direct and no port binding\n6df75ca696 Deprecate allow_stateless_action_supported config option\n5250598c80 Improve scheduling L3/DHCP agents, missing lower binding\nindexes\n8f7cc690a9 Remove ``L3Scheduler._bind_routers`` method\n47fef55e25 Add a method to retrieve and register \"rpc_workers\" config\nknob\n359030f741 Do not log full network on reload_allocations\n4e71675ddb rbacs: add functionl test that asserts behavior on\nnetworks/users\n5c697b8d60 allow manila ports to do multiple port binding for ML2\nc15313ea21 Removed duplicated job call in experimental queue\nc93c31852b Adapt deploy_rootwrap filters path for tox4\n61b2917a3e Filter subnets by \"enable_dhcp\" flag using the correct type\n8be42f6d72 Enable qos extension_driver in ovn_db_sync\n3c2af924d9 Add CI jobs using SQLAlchemy master branch\nabfa8940b1 [OVN] Fix the OVN Agent config file location\n54eff20a72 [OVN] New OVN Neutron Agent extension: QoS for HWOL\n3cf4899cf0 [OVN] Adding support for VNIC type virtio-forwarder.\n07fd743dc9 Remove extra call to config_logging() for neutron-server\ne98a268de4 Propose replacement of ORM from_self()\n1c27d8b5b4 [Secure RBAC] Fix policy to get flavors\n17be2505d1 [Secure RBAC] Allow network owner to delete ports plugged to\nthe network\n5d505e7b7c [Secure RBAC] Add shared_security_groups rule\n221cf9497f [Secure RBAC] Add shared_qos_policy rule\nf7e31b4c05 [OVN] Allow logging all traffic related to an ACL\n267ad86291 Add tox Python3.11 job to the testing queues\n83da1e6d79 Improve message for subnet gateway out of host IP addresses\nrange\n7cb481a3dc Allow disable stateful security group extension on older OVN\n48c3621292 Imported Translations from Zanata\n3727103ce7 Fix get_link_devices() with index argument\n67e616b238 Never raise an exception in notify()\nd0c7bb653a [OVN] Implementation of OVN Neutron Agent\nc609084b59 Add doc note on nf_conntrack module requirement\n0ae432e717 Fullstack: Wait placement process fixtrue to really stop\n910574a5cc Use SLQAlchemy ORM \"relationship\" instead of \"relation\"\n1d22cfa99a Change flag check order in wait_until_address_ready()\ned68ba4a4c Do not ignore attributes in bulk port create\nd73f75c551 [API] Add API extension and definition for default SG rules\n1a3bdff18a Honor debug mode in keepalived-state-change script logs\nec438577bd Use dhcpcd client in the tempest slow jobs\nf235ce8791 Add new debugging logs for ``ARPSpoofTestCase``\n78c3d52bb4 [OVN] Introduce the new OVN Neutron Agent\nfeb9f2b21c ovn-migration: Stop neutron server while running db sync\n2495ad2856 Avoid failure of get_device_details if net from cache is\nempty\nb6750fb2b8 [OVN] Add vnic_type and binding profile capabilities to LSP\ninfo\na3bf74baa5 Ignore reno artefacts (RELEASENOTES.rst and reno.cache)\n1d84aaf877 Bump the minimum pyroute2 version to 0.7.3\n463c3df4cf [ovn] Do not create empty default route when empty gateway_ip\n70a86637e7 Allow shared resources between physical and tunnelled\nnetworks\nf28e416599 Add 3 secs to wait for keepalived state change\nd7f68a0ce7 Set proxy_delay to one when using proxy ARP\nd0e64c6617 Add auto-allocated-topology delete example to docs\n3354b43d5a Increase fullstack job\u0027s timeout\ne2097157bb Allow multiple IPv6 ports on router from same network\nml2/ovs+dvr\nde0918cea3 Add debug logs to OVSInterfaceDriver.plug_new\n7751ac32a7 Add info on segments extension to contrib guide\ned048638f4 Add text to WSGI config guide for debugging\nc362016620 Ensure that MAC address of the device is set correctly\n3e1e2d63b3 Enable rpc notifications only when rpc_workers \u003e\u003d 1\n1c89db495c [OVN][FT] Also update nb_cfg along with nb_cfg_timestamp\nd59bf5ff6c Temporary update envlist for tox4\n25ec6e7e4f Set ip_nonlocal_bind to 1 for HA routers and DVR snat\n07fd0e101b Revert \"Documentation for adding extra port conflict fix\"\n6501ebc6ca Documentation for adding extra port conflict fix\nf759915ab0 Fix default value for MTUs, when not provided\n02030f037a ovs-dpdk support in neutron-sanity-check\n9f9478071e Bump Cirros version used in CI jobs to 0.6.1\nbee07defac Track all interfaces in Keepalived\n26da863ca6 Remove deprecated opt keepalived_use_no_track\n4fa85b9a99 Fix incorrect mysql examples\nedf48e46a1 Improve agent provision performance for large networks\n5e19eaadbb [OVN] Fix availability zones changes check\n3ebdfe612a Support for minimum bandwidth rules in tunnelled networks\n572185e1fd Delete the PB level registers when deleting the duplicated PB\nbe0776d647 Add sanity check for keepalived gar on SIGHUP support\nb040e6139e [Fullstack] use common assert for ncat connection\n3d575f8bd0 Add an env variable \"PROCESS_TAG\" in ``ProcessManager``\n08e485d0cd Ignore python warnings in the neutron-status functional tests\n370d8bcea3 [Fullstack] Wait 10 seconds to ensure that MAC address is\nconfigured\n4edf94041f Remove note about migration from lib/neutron-legacy to\nlib/neutron\n0475a3e3f3 Fix typos in routed networks doc file\ndad23fdcdb Strictly delete arp_spoofing_protection flows\nfaa4803302 dhcp: fix issue when network is already removed\nbe0996c308 segment: enable multisegments support for host\n7c449f1833 Enable ML2/OVS backend in the -ovs- grenade jobs explicitly\n9401da6c7c Handle if \"TRUNK_SKELETON\" is not loaded\nb5fdad3cfb Pin OVS_BRANCH in ipv6 ovn job\n14d0ff0317 Add ``devlink.get_port`` method for devlink ports\n0d315d2467 Update .pylintrc\n86badcfe2d Fix some pylint indentation warnings\n4e75c69ab6 Add extra logs when checking if FIP GW port has been created\ncbd03c7ea5 Wait until local OVS DB updates Bridge \"datapath_id\"\n1d46b19951 Pin OVS_BRANCH to master again\n4cc611d319 Fix handling the restart of ovn-controllers\n44ad4bb517 Remove deprecated \"list_moved_globals\" script\n8bcaf1632f Limit tox version to \u003c4\nc2fb0b16b6 Rename setup-mtu-t1 argument because it is misleading\n135daae62d Use same ovs/ovn versions in local test setups\naaae079883 Fix homedir permissions\n405414772e Fix tox.ini for tox4\nca73925f42 Pin OVS_BRANCH to working commit\nf7c0f466be Fix text in OVN RevisionConflict exception to match code\nc7cd387796 Amend documentation for fullstack tests\ndfe69472a8 Check if port exists in ``update_port_virtual_type`` method\n2fd17aa876 [OVN] Initialize the mech driver variable in\n``OVNL3RouterPlugin``\n06fad386ba Fix firewall_driver examples in docs\n74224e79e0 Gracefully restart dnsmasq to not break tcp DNS\nf22aa5dfdd Discard port with ofport -1 in _get_ofport_moves\n81980146cb Add and delete multiple ip addresses in one priv call\n92efd8e45b Dont raise RouterInterfaceNotFound on overlap check router\nports\n49eab7d308 Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl\ninit\n234c5a376b [sqlalchemy-20] Add missing DB context decorator\n88b7db0e0c Mock ls_get in OVN metadata delete network unit test\na8adb5aa9c [OVN] Implement abstract \"post_connect\" in\n\"MetadataAgentOvnSbIdl\"\n81330921d8 Fix some pylint indentation warnings\ncca0b3a026 Fix some pylint indentation warnings\nfd423f7b1b Reset the MeterIDGenerator singleton after each test\n4416706931 Use neutron-lib constants for OVS table numbers\n26c55efe34 Do not emit ssl warnings in the test_ssl_connection test\nf7dd7790f5 Fix bulk create without mac\n895f818cb3 Do not mock the hash method for testing\n\"get_vlan_device_name\"\n5f819c0767 Update neutron-tox-override-jobs template\nc813b658d0 Fix duplicated routes exceptions\nf629b77d3c Fix behaviour of enable/disable in OVN network log\n5dad500481 Remove deprecation warning of \"ovs-ofctl\" usage\n2272b9be95 Add address scope to the OVN LSP port registers\n499ae95881 Bump pecan to fix #1829042\n7a7448ead4 Configure placement credentials for segments plugin\n6d8ada0ac9 [S-RBAC] Allow admin user to do all API requests by default\n4d1a7bd0bc neutron-dynamic-routing now works with OVN\n3b1fd52b41 Exclude neutron-dynamic-routing from abandon script\ne1d70f9cb0 Disable \"test_live_migration_with_trunk\" execution from dvr\njob\neeff5b3c81 db: add segment_index to the unique constraint\n67bd2badc9 Fix missing [designate] options\n631ad0ce6b ovn: use dicts instead of sets of tuples\nae845d572c ovn: inline ensure_all_networks_provisioned method\n686698284b Update tap ip in metadata agent when metadata port ip updated\n1a9ae924fd Add VXLAN support reference in the OVN installation guide\n3093aaab13 ovn: first tear down old metadata namespaces, then deploy new\n0ef4f98825 Remove policy rule for get_network:router:external\n2081910d6d Always create a \"router_extra_attributes\" register per router\n5b64ac9013 Use the neutron-lib constant for PACKET_RATE_LIMIT table\nnumber (2)\n9a483f02be Make MechanismDriverContext plugin_context public\n33de608f04 Avoid race condition when deleting trunk bridges\nc5640eee11 Use the neutron-lib constant for PACKET_RATE_LIMIT table\nnumber\nbf44e70db6 OVN: Add support for DHCP option \"domain-search\" for IPv4\n25d4e3862b Code move for metadata haproxy\nc3ebefa5f7 Refactor for meter ID Generator\nff66ca9275 Execute functional test \"test_get_all_devices\" alone\n51827d8e78 Remove \"test_live_migration_with_trunk\" execution\na2090e4cbc Doc: Capitalize keystone domain names\nc5ee9f3495 Load the required configuration options in the UT classes\nbb2d706107 [doc] Metadata service supports IPv6 networks\nb7d04d5d92 Code move for metadata signature function\nc6a83988ea Replace Linux Bridge with Open vSwitch in installation docs\n81004125c4 [OVN] Prevent maintenance task fail resource stdattr does not\nexist\n9e2a0ac058 Use --exclude-regex in tox.ini\n609ebd9504 test: fix docstring validation function\n562e9704f8 Update documentation link for openSUSE index.\nb1714a2b9d Fix some pylint indentation warnings\n55b16d7b7c Fix some pylint indentation warnings\n3afa1b1471 Fix some pylint indentation warnings\n3702575bfe Fix some pylint indentation warnings\nf77c7c9584 Fix some pylint indentation warnings\n291eabb0b6 Fix some pylint indentation warnings\nd1c2d2c4fe dhcp: support multiple segmentations per network\n01b11df1b5 Add tempest job which will run on the oldest Ubuntu release\n0da328fbee Add \"tempest-integrated-networking\" job to the check and gate\nqueues\n1abb77d7a6 Check subnet overlapping after add router interface\n304394ed77 Revert \"Temporary make rally job non voting\"\nc871135cdd Remove xenapi from neutron ml2 config opts\n8fcf00a36d Disable in-band management for bridges before setting up\ncontrollers\nc33b47edc7 Do not keep gateway port when notifying for router update\n7c1a894ce5 Nit: network_update in ovs_neutron_agent has a bad LOG\n963f8e30d8 CI: Add tick-tick grenade jobs to check queue\nda335193ef Promote neutron-ovn-rally-task job to be voting in check\nqueue\n8d5985ba6a [Docs] Remove python version from the ci jobs summary\n99339a9704 Clean \"ovn ipv6 only\" jobs templates\nb7b8f7c571 [OVN] Avoid deadlock when cleaning hash ring nodes\ncb332acb29 dhcp: add/use cleanup stale devices API\n1d8e3b79db dhcp: make device manager to clean only not used devices\n4551718bad Add \"neutron-ovn-tempest-ovs-release\" job to the check and\ngate queue\nee7bfca937 Remove unnecessary bullet points, unify formatting.\n7a5a61616a Remove reference to \u0027all-plugin\u0027 tox environment\n72dd768a12 Revert \"Since OVN 20.06, config is stored in\n\"Chassis.other_config\"\"\n628442aed7 Spread OVN metadata agent heartbeat response in time\nca2e593177 Change grenade job names suffix to \"skip-level\"\n240f2c6aeb Split Hash Ring probing from the maintenance task\n1920f2a4e3 Add new oslo master CI jobs to the periodic queue\n7b8b797dea Load the required configuration options in the UT classes\n4858315286 Fix misplaced comparison constant warnings\n3d0f1ef4fa Remove some unnecessary list comprehensions\n70a2029ca1 [OVN] Set the default OVN metadata worker number to 0\n34b81ae9b7 Remove OVN gap minimum bandwidth support\n229d798a45 [OVN] Allow only one physical network per bridge\nf43891bf86 [OVN] Allow to execute ``MetadataProxyHandler`` in a local\nthread\n50eee19723 ovn: Use ovsdb-client to create neutron_pg_drop\n536498a29a Since OVN 20.06, config is stored in \"Chassis.other_config\"\n0c9e7d4e18 Update grenade skip level jobs for new release\nde18dd2c2e [ovn migration] Use ecsda ssh key instead of rsa\nd7b44f7218 Allow to pass EUI64 IP address as fixed ip for the port\na3e68e8f76 Fix ipam_pluggable_backend unit tests module\na8a43c8045 Execute \"IpMonitorTestCase\" tests always inside a namespace\nf8fa909444 Since OVN 20.06, config is stored in \"Chassis.other_config\"\n7b15ad9433 Bump minimum OVN version to v21.06.0\ne4cc40f114 fix: Fix url of Floodlight\n21491efd9f Port provisioning should retry only for VM ports\n178ee6fd3d [Trunk] Update the trunk status with the parent status\n0b2844b8a8 Add .pylintrc to irrelevant-files for non pep8 jobs\n72493d1472 Add missing comma in .pylintrc\n3f20cabc4f Revert \"Move the \"ovn_hash_ring\" clean up to maintenance\nworker\"\n56f89131dd docs: Remove errant indentation, mark up literals\n2f0919f7df Accept a port deletion with missing port binding information\n1fe4edab80 Open Antelope DB branch\n41a7479d0f Handle several dhcp agents for metadata over ipv6\n819a1bb3e6 Move the \"ovn_hash_ring\" clean up to maintenance worker\ne9b5a27001 Use \"OVNMechDriver\" instance in\n\"TestOvn[Nb|Sb]IdlNotifyHandler\"\n9eddb132b2 Fix openstack-tox-py39-with-oslo-master job\na765b7acbb [sqlalchemy-20] Add py39 oslo master to experimental\nea11928ba0 Neutron needs oslo.config \u003e\u003d 9\n9a1b5d1e51 Imported Translations from Zanata\n46bc1e7f13 Switch to 2023.1 Python3 unit tests and generic template name\n4914d7fab2 Update master for stable/zed\n639257c77b Make path for tcpdump release-specific\ncca48be73d Imported Translations from Zanata\nf7578e51f2 Collect ovn/ovs logs in singlenode jobs\n9655466763 [OVN] Rate limit the \"Disallow caching\" log from hash ring\n\nChange-Id: Iea7bef0e3e3915bea6859dabccdd032bba6717ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0693572151c1343be30fbdc0d1b2f5df2aa537e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0693572151c1343be30fbdc0d1b2f5df2aa537e5"}]},"branch":"refs/heads/master"},"7c1e7fefc44deb39a98d0e5c820c94dea23d88a4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-10 11:22:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/19/882819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/882819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/882819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/882819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/882819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/882819/2"}}},"commit":{"parents":[{"commit":"8e7f1505afa2cf5e0be35a0e460d3367173233bf","subject":"Merge \"[octavia] Transition Xena to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e7f1505afa2cf5e0be35a0e460d3367173233bf"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-09 21:28:03.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-05-10 11:22:37.000000000","tz":0},"subject":"Release Neutron 23.0.0.0b1 (Neutron Bobcat beta 1)","message":"Release Neutron 23.0.0.0b1 (Neutron Bobcat beta 1)\n\nThis release contains the change [1] that enables by default\nthe new sRBAC policies.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/879827\n\n$ git log --oneline --no-merges 21.0.0.0rc1..9319ba00a\nc3602ac19b [OVN] Update ovn meter when neutron server reloads\nc0af5b3b5e Reduce lock contention on subnets\n256297fc7f rbacs: clean-up to use defined constants ACCESS_*\n6eb7006801 Drop retries in tests for TimeoutException\n30c0e5699e Fix doc links for networking option 2\nb1cc242fad Add a method to retrieve router gateway ports\n7a59cf0eb9 Correct planned removal for maintenance function\n2be53b1719 [functional] Fix db_set use in test_cascading_del_in_txn\na9963e90d9 ``_get_ovn_version`` returns a 3 element tuple\n4ad979a534 Mark \"test_port_creation_and_deletion\" as unstable\n9ac59e4b4a Avoid retrieving ports if network list is empty\n4bac350f68 Remove \"neutron-ovn-tempest-ovs-release-ubuntu-old\" job\nb19b55909d Don\u0027t set and remove immediately DEAD VLAN tag in tests\n88ce859b56 Change API to validate network MTU minimums\n9f22dc1d3a Doc: Add FWaaS v2 install details\n853dc2570d Add py39 jobs to tox override template\n39d252da7c Deprecated support for Windows OS\n82029c2c51 Use a writer context for the online alembic migrations\n2cafe0a9c4 Revert \"Move to python3.9 as minimal python version\"\nf93c9be1c1 Move to python3.9 as minimal python version\n4e27e27ae2 Replace context decorators with context managers\ndd184c5c10 Fix Loki tempest jobs\nb31453af47 [OVN] Admin procedure for duplicated or deleted OVN agents\n872b53f618 Fix functional tests modules which are using PluginFixture\n670cc383e0 [S-RBAC] Switch to new policies by default\nd757c530bc Update QoS documentation\na84567b8d6 Remove the ``OVNSqlFixture`` class workaround\n05ba4257de Remove creation of DHCP port in the OVN metadata unit test\n43829301f3 Handle \"no such process\" during keepalived process cleanup\n34ea8988d6 [OVN] Add update event to ``OVSInterfaceEvent`` class\na5e26408d2 Fix dns_integration and ml2 plugin unit tests modules\n5db17654b9 Fix servicetype unit tests module\nf92d6fa72a Fix segments unit tests module\n6b5acb5835 [S-RBAC] Get availability zone API available for READER role\n340fd3cfe7 Run master jobs only on master branch\n706a0e0268 Fix parent of neutron-ovn-tempest-with-uwsgi-loki\n28961c8b76 Fix network_segment_range unit tests module\n73ac4510c7 [grenade] Collect ovn services logs\n2364327bd1 Run periodic jobs in experimental queue too\n9d51633013 Fix intermittent failures in finding metada port in SB DB\n267efd2984 OVN: Always try and create a metadata port on subnets\nc7ef824941 Do not check the context object in ``TestMeteringPlugin``\n5510cdab92 [ovn] OVNClient._get_router_ports: Drop unused parameter\ne5d4499672 [ovn] Drop use of LR OVN_GW_NETWORK_EXT_ID_KEY\nd67d1c2736 [ovn] Drop use of OVN_GW_PORT_EXT_ID_KEY\n69f30c92ef [sqlalchemy-20] Add reader context to\n``get_ports_on_host_by_subnet``\n7dfbdf65a7 Add support for localnet_learn_fdb OVN option\ne374b82d8f [CI][fullstack/functional] Report slowest tests\n8eecccfeae [S-RBAC] Allow network owners to get ports from that network\n95c19c8868 Checkout \"sqlalchemy/alembic\" main branch in\nsqlalchmey-master jobs\n3de8ebebd8 Pin OVS_BRANCH to working commit\na06fe7cfd9 Fix NoSuchOptError error in Ipam unit tests\n5a17f2b24a Pass physical bridge informations to OVS agent extension API\n7828acaf4f [ovn] Add end to end test for QosExtension\n2aee961ab6 Suppress IPv6 metadata DAD failure and delete address\nade2a9f893 Mark \"test_multiple_agents_for_network\" as unstable\nfa172ab7dc rbacs: fix typo, s/cxt/ctx\ne6de524555 rbacs: filter out model that are already owned by context\n7073410be3 Bump skip-level lower version to stable/zed\n775d5de9f1 Add debug information to\n``MacvtapAgentTestCase.test_get_all_devices``\n0ec04dd638 Ensure redirect-type\u003dbridged not used for geneve networks\nd4654e3011 Filter out unsatisfied routers in SQL\n33c4a2d97e Update url and package name\n3cc28a004a Add plugin.spec to irrelevant-files\nb777aa57b2 Update the quota guide examples\ncfe38a0014 Change name of ``_TestIsSessionActive`` test case.\n27b3eacd3d Revert \"Ensure vlan network traffic is not centralized\"\ne4da60740b [sqlalchemy-20] Do not use strings for aatribute names in\nloader options\n0a69dd5e3d [sqlalchemy-20] query.join should define one table/column per\ncall\n4b8e484e1d Increase port name size and type to internal\n2ccf1e1e90 [OVN] OVN agent should register \"Chassis_Private\" by default\nf42f1cfa69 [sqlalchemy-20] Provide SQL \"case\" expression correct input\nparemeters\n831ac3152d Fix a number of configuration typos\n5b4ed5b117 Fix concurrent port binding activate\n1646e5b28d Add neutron-lib project to the SQLAlchemy master branch CI\njobs\n0220236c63 Cleanup before executing \"test_get_all_devices\"\n6e1dbe9781 Add oslo.db project to the SQLAlchemy master branch CI jobs\n97aa84b69a Open the 2023.2 (Bobcat) DB branch\nc97dcfd03f doc: state that O flag can be 0 in dhcpv6-stateful\n6358495720 Delete sg rule which remote is the deleted sg\n04d3f889ef Fix metadata agent intermittent test failures\n9704dca84e [OVN] Explicitly define the fixed IPs for the metadata port\ndea48cfc0a Only create a frozen  Row on matching events\n711fbb9820 Imported Translations from Zanata\n8ea2a9c128 Fix typo in unit test\n442b437a81 Change external process manager tests to clean temp files\nefab60c0bf Try to optimize Mysql server mem usage on some CI jobs\n745497a112 [OVN] Remove \"update_port_qos_with_external_ids_reference\"\n5c98d9e8d1 [OVS] Parse the \"permitted_ethertypes\" at the FW\ninitialization\nbffa642b35 [OVN] Method ``get_port_qos`` should always return 2 values\n008277b8c1 [OVS] Allow custom ethertype traffic in the ingress table\nb9567033fc [OVN] Use the BW values retrieved from ``get_port_qos``\n08eb8e2498 Use neutron-lib policy rules\nc8ccf2ffbb [OVN] Change oslo config options entry point for the OVN\nagent\ncf96bd8bdf ovs: fix regression when vlan mapping is not already\nregistered\nd44f164f4d ovn_idl_impl: fix a logic bug in get_sg_port_groups\nd9358b67bd functional: set dns_domain config option after its\nregistration\n8946684fb2 Remove duplicate rows in MySQL query output\n5d2086c698 Add 2023.1 release name in routed networks doc\nb6bc4c8a66 Add Lajos Katona to Client and Doc areas as lieutenant\nd4a85833a7 [sqlalchemy-20] The Session.begin.subtransactions flag is\ndeprecated\n0a214b0437 Imported Translations from Zanata\n999116126e Add full support for OVN NB \"Gateway_Chassis\" table\n4254ccd1bc Update master for stable/2023.1\n8e3bddbf8b Ensure vlan network traffic is not centralized\n39b65575cd Change the release tag to use the release identification\n44ac03de14 Fix policy unit test deprecation warnings\nebcde41fc8 [OVN] Add ``get_gateway_chassis_az_hints`` method to OVN API\n41846a6e8b [sqlalchemy-20] The Connection.connect() method is considered\nlegacy\n2af5fd889b Add sleep before checking if ovs port is in the namespace\n12093015de Add Jens Harbott as Lieutenants in Infra area\n19482a049b Add Slawomir Kaplonski as Lieutenant in API, Client and\nTesting areas\n51005388b8 Make retrieval of port mac column safe\n6cac935824 Update lieutenants for ovn-octavia-provider\n948c9e02e3 [S-RBAC] Add release note about full support for new policies\n289ae97c1c Add missing osprofiler options\nacb809eea4 [OVN] Ensure traffic for provider vlan networks is not\ntunneled\n7f46ea9b6d Remove unused rbac rule\n75e8360224 Reintroduce agent bridge resync test\nc2a02d3c69 Remove two duplicated unit tests\ndbfc18d1fa Apply Ironic\u0027s server-ip-address as TFTP next-server\nbdf0c8bdad Move register_common_config_options call to DietTestCase\ne7c92d92e4 [sqlalchemy-20] Use the .begin() method of Engine\n827fbd01c3 Normalise format of OVN agent heartbeat timestamp\ne68e4162ce Prevent router_ha_interface port from being removed via API\naa40aef70f Mark fullstack TestMultiSegs.test_multi_segs_network as\nunstable\n169ef05589 Change neutron-ovs-tempest-dvr-ha-multinode-full job\u0027s config\nb00ae45140 ml2 plugin: use const from neutron-lib\nf991fc2fbe [ovn][ipv6] Skip test_update_delete_extra_route\n1e41558851 Reduce number of routers created in\n``test_gateway_chassis_rebalance``\n51d449babf Move tox-py311 job override into job-templates.yaml\nf5dc708e1a Check port.tag is not DEAD_VLAN_TAG in\n``DHCPAgentOVSTestFramework``\nc61ce3447e Format correctly (dialect\u003dmac_unix_expanded) the MAC\naddresses\n502c0253de [ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class\ne9ce025a69 [dhcp agent] Fix clean_devices\nd052da85cf Mark ``test_port_creation_and_deletion`` as unstable\n5e15c4a4d4 Allow other kwargs in ``DhcpBase.disable``\n5dac5f8c6c [ovn-ipv6] Temporary exclude test_hotplug_nic\n67a232fe71 Remove tripleo related jobs from neutron queues\n595c387efb [UT] Mock the retry decorator when creating duplicated\nNDPProxy\nf9880c2013 Increase tox-py311 timeout to 3600 seconds\n24152f2019 Add debugging exception messages in\n``test_ovs_and_ovs_events``\nfce516e346 [OVN] Bump the port revision number in trunk driver\nd83e243c6d [OVN] Add missing LSP device_owner info in trunk driver\n76f3b75ad6 [OVN] Skip ports with VNIC type direct and no port binding\n6df75ca696 Deprecate allow_stateless_action_supported config option\n5250598c80 Improve scheduling L3/DHCP agents, missing lower binding\nindexes\n8f7cc690a9 Remove ``L3Scheduler._bind_routers`` method\n47fef55e25 Add a method to retrieve and register \"rpc_workers\" config\nknob\n359030f741 Do not log full network on reload_allocations\n4e71675ddb rbacs: add functionl test that asserts behavior on\nnetworks/users\n5c697b8d60 allow manila ports to do multiple port binding for ML2\nc15313ea21 Removed duplicated job call in experimental queue\nc93c31852b Adapt deploy_rootwrap filters path for tox4\n61b2917a3e Filter subnets by \"enable_dhcp\" flag using the correct type\n8be42f6d72 Enable qos extension_driver in ovn_db_sync\n3c2af924d9 Add CI jobs using SQLAlchemy master branch\nabfa8940b1 [OVN] Fix the OVN Agent config file location\n54eff20a72 [OVN] New OVN Neutron Agent extension: QoS for HWOL\n3cf4899cf0 [OVN] Adding support for VNIC type virtio-forwarder.\n07fd743dc9 Remove extra call to config_logging() for neutron-server\ne98a268de4 Propose replacement of ORM from_self()\n1c27d8b5b4 [Secure RBAC] Fix policy to get flavors\n17be2505d1 [Secure RBAC] Allow network owner to delete ports plugged to\nthe network\n5d505e7b7c [Secure RBAC] Add shared_security_groups rule\n221cf9497f [Secure RBAC] Add shared_qos_policy rule\nf7e31b4c05 [OVN] Allow logging all traffic related to an ACL\n267ad86291 Add tox Python3.11 job to the testing queues\n83da1e6d79 Improve message for subnet gateway out of host IP addresses\nrange\n7cb481a3dc Allow disable stateful security group extension on older OVN\n48c3621292 Imported Translations from Zanata\n3727103ce7 Fix get_link_devices() with index argument\n67e616b238 Never raise an exception in notify()\nd0c7bb653a [OVN] Implementation of OVN Neutron Agent\nc609084b59 Add doc note on nf_conntrack module requirement\n0ae432e717 Fullstack: Wait placement process fixtrue to really stop\n910574a5cc Use SLQAlchemy ORM \"relationship\" instead of \"relation\"\n1d22cfa99a Change flag check order in wait_until_address_ready()\ned68ba4a4c Do not ignore attributes in bulk port create\nd73f75c551 [API] Add API extension and definition for default SG rules\n1a3bdff18a Honor debug mode in keepalived-state-change script logs\nec438577bd Use dhcpcd client in the tempest slow jobs\nf235ce8791 Add new debugging logs for ``ARPSpoofTestCase``\n78c3d52bb4 [OVN] Introduce the new OVN Neutron Agent\nfeb9f2b21c ovn-migration: Stop neutron server while running db sync\n2495ad2856 Avoid failure of get_device_details if net from cache is\nempty\nb6750fb2b8 [OVN] Add vnic_type and binding profile capabilities to LSP\ninfo\na3bf74baa5 Ignore reno artefacts (RELEASENOTES.rst and reno.cache)\n1d84aaf877 Bump the minimum pyroute2 version to 0.7.3\n463c3df4cf [ovn] Do not create empty default route when empty gateway_ip\n70a86637e7 Allow shared resources between physical and tunnelled\nnetworks\nf28e416599 Add 3 secs to wait for keepalived state change\nd7f68a0ce7 Set proxy_delay to one when using proxy ARP\nd0e64c6617 Add auto-allocated-topology delete example to docs\n3354b43d5a Increase fullstack job\u0027s timeout\ne2097157bb Allow multiple IPv6 ports on router from same network\nml2/ovs+dvr\nde0918cea3 Add debug logs to OVSInterfaceDriver.plug_new\n7751ac32a7 Add info on segments extension to contrib guide\ned048638f4 Add text to WSGI config guide for debugging\nc362016620 Ensure that MAC address of the device is set correctly\n3e1e2d63b3 Enable rpc notifications only when rpc_workers \u003e\u003d 1\n1c89db495c [OVN][FT] Also update nb_cfg along with nb_cfg_timestamp\nd59bf5ff6c Temporary update envlist for tox4\n25ec6e7e4f Set ip_nonlocal_bind to 1 for HA routers and DVR snat\n07fd0e101b Revert \"Documentation for adding extra port conflict fix\"\n6501ebc6ca Documentation for adding extra port conflict fix\nf759915ab0 Fix default value for MTUs, when not provided\n02030f037a ovs-dpdk support in neutron-sanity-check\n9f9478071e Bump Cirros version used in CI jobs to 0.6.1\nbee07defac Track all interfaces in Keepalived\n26da863ca6 Remove deprecated opt keepalived_use_no_track\n4fa85b9a99 Fix incorrect mysql examples\nedf48e46a1 Improve agent provision performance for large networks\n5e19eaadbb [OVN] Fix availability zones changes check\n3ebdfe612a Support for minimum bandwidth rules in tunnelled networks\n572185e1fd Delete the PB level registers when deleting the duplicated PB\nbe0776d647 Add sanity check for keepalived gar on SIGHUP support\nb040e6139e [Fullstack] use common assert for ncat connection\n3d575f8bd0 Add an env variable \"PROCESS_TAG\" in ``ProcessManager``\n08e485d0cd Ignore python warnings in the neutron-status functional tests\n370d8bcea3 [Fullstack] Wait 10 seconds to ensure that MAC address is\nconfigured\n4edf94041f Remove note about migration from lib/neutron-legacy to\nlib/neutron\n0475a3e3f3 Fix typos in routed networks doc file\ndad23fdcdb Strictly delete arp_spoofing_protection flows\nfaa4803302 dhcp: fix issue when network is already removed\nbe0996c308 segment: enable multisegments support for host\n7c449f1833 Enable ML2/OVS backend in the -ovs- grenade jobs explicitly\n9401da6c7c Handle if \"TRUNK_SKELETON\" is not loaded\nb5fdad3cfb Pin OVS_BRANCH in ipv6 ovn job\n14d0ff0317 Add ``devlink.get_port`` method for devlink ports\n0d315d2467 Update .pylintrc\n86badcfe2d Fix some pylint indentation warnings\n4e75c69ab6 Add extra logs when checking if FIP GW port has been created\ncbd03c7ea5 Wait until local OVS DB updates Bridge \"datapath_id\"\n1d46b19951 Pin OVS_BRANCH to master again\n4cc611d319 Fix handling the restart of ovn-controllers\n44ad4bb517 Remove deprecated \"list_moved_globals\" script\n8bcaf1632f Limit tox version to \u003c4\nc2fb0b16b6 Rename setup-mtu-t1 argument because it is misleading\n135daae62d Use same ovs/ovn versions in local test setups\naaae079883 Fix homedir permissions\n405414772e Fix tox.ini for tox4\nca73925f42 Pin OVS_BRANCH to working commit\nf7c0f466be Fix text in OVN RevisionConflict exception to match code\nc7cd387796 Amend documentation for fullstack tests\ndfe69472a8 Check if port exists in ``update_port_virtual_type`` method\n2fd17aa876 [OVN] Initialize the mech driver variable in\n``OVNL3RouterPlugin``\n06fad386ba Fix firewall_driver examples in docs\n74224e79e0 Gracefully restart dnsmasq to not break tcp DNS\nf22aa5dfdd Discard port with ofport -1 in _get_ofport_moves\n81980146cb Add and delete multiple ip addresses in one priv call\n92efd8e45b Dont raise RouterInterfaceNotFound on overlap check router\nports\n49eab7d308 Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl\ninit\n234c5a376b [sqlalchemy-20] Add missing DB context decorator\n88b7db0e0c Mock ls_get in OVN metadata delete network unit test\na8adb5aa9c [OVN] Implement abstract \"post_connect\" in\n\"MetadataAgentOvnSbIdl\"\n81330921d8 Fix some pylint indentation warnings\ncca0b3a026 Fix some pylint indentation warnings\nfd423f7b1b Reset the MeterIDGenerator singleton after each test\n4416706931 Use neutron-lib constants for OVS table numbers\n26c55efe34 Do not emit ssl warnings in the test_ssl_connection test\nf7dd7790f5 Fix bulk create without mac\n895f818cb3 Do not mock the hash method for testing\n\"get_vlan_device_name\"\n5f819c0767 Update neutron-tox-override-jobs template\nc813b658d0 Fix duplicated routes exceptions\nf629b77d3c Fix behaviour of enable/disable in OVN network log\n5dad500481 Remove deprecation warning of \"ovs-ofctl\" usage\n2272b9be95 Add address scope to the OVN LSP port registers\n499ae95881 Bump pecan to fix #1829042\n7a7448ead4 Configure placement credentials for segments plugin\n6d8ada0ac9 [S-RBAC] Allow admin user to do all API requests by default\n4d1a7bd0bc neutron-dynamic-routing now works with OVN\n3b1fd52b41 Exclude neutron-dynamic-routing from abandon script\ne1d70f9cb0 Disable \"test_live_migration_with_trunk\" execution from dvr\njob\neeff5b3c81 db: add segment_index to the unique constraint\n67bd2badc9 Fix missing [designate] options\n631ad0ce6b ovn: use dicts instead of sets of tuples\nae845d572c ovn: inline ensure_all_networks_provisioned method\n686698284b Update tap ip in metadata agent when metadata port ip updated\n1a9ae924fd Add VXLAN support reference in the OVN installation guide\n3093aaab13 ovn: first tear down old metadata namespaces, then deploy new\n0ef4f98825 Remove policy rule for get_network:router:external\n2081910d6d Always create a \"router_extra_attributes\" register per router\n5b64ac9013 Use the neutron-lib constant for PACKET_RATE_LIMIT table\nnumber (2)\n9a483f02be Make MechanismDriverContext plugin_context public\n33de608f04 Avoid race condition when deleting trunk bridges\nc5640eee11 Use the neutron-lib constant for PACKET_RATE_LIMIT table\nnumber\nbf44e70db6 OVN: Add support for DHCP option \"domain-search\" for IPv4\n25d4e3862b Code move for metadata haproxy\nc3ebefa5f7 Refactor for meter ID Generator\nff66ca9275 Execute functional test \"test_get_all_devices\" alone\n51827d8e78 Remove \"test_live_migration_with_trunk\" execution\na2090e4cbc Doc: Capitalize keystone domain names\nc5ee9f3495 Load the required configuration options in the UT classes\nbb2d706107 [doc] Metadata service supports IPv6 networks\nb7d04d5d92 Code move for metadata signature function\nc6a83988ea Replace Linux Bridge with Open vSwitch in installation docs\n81004125c4 [OVN] Prevent maintenance task fail resource stdattr does not\nexist\n9e2a0ac058 Use --exclude-regex in tox.ini\n609ebd9504 test: fix docstring validation function\n562e9704f8 Update documentation link for openSUSE index.\nb1714a2b9d Fix some pylint indentation warnings\n55b16d7b7c Fix some pylint indentation warnings\n3afa1b1471 Fix some pylint indentation warnings\n3702575bfe Fix some pylint indentation warnings\nf77c7c9584 Fix some pylint indentation warnings\n291eabb0b6 Fix some pylint indentation warnings\nd1c2d2c4fe dhcp: support multiple segmentations per network\n01b11df1b5 Add tempest job which will run on the oldest Ubuntu release\n0da328fbee Add \"tempest-integrated-networking\" job to the check and gate\nqueues\n1abb77d7a6 Check subnet overlapping after add router interface\n304394ed77 Revert \"Temporary make rally job non voting\"\nc871135cdd Remove xenapi from neutron ml2 config opts\n8fcf00a36d Disable in-band management for bridges before setting up\ncontrollers\nc33b47edc7 Do not keep gateway port when notifying for router update\n7c1a894ce5 Nit: network_update in ovs_neutron_agent has a bad LOG\n963f8e30d8 CI: Add tick-tick grenade jobs to check queue\nda335193ef Promote neutron-ovn-rally-task job to be voting in check\nqueue\n8d5985ba6a [Docs] Remove python version from the ci jobs summary\n99339a9704 Clean \"ovn ipv6 only\" jobs templates\nb7b8f7c571 [OVN] Avoid deadlock when cleaning hash ring nodes\ncb332acb29 dhcp: add/use cleanup stale devices API\n1d8e3b79db dhcp: make device manager to clean only not used devices\n4551718bad Add \"neutron-ovn-tempest-ovs-release\" job to the check and\ngate queue\nee7bfca937 Remove unnecessary bullet points, unify formatting.\n7a5a61616a Remove reference to \u0027all-plugin\u0027 tox environment\n72dd768a12 Revert \"Since OVN 20.06, config is stored in\n\"Chassis.other_config\"\"\n628442aed7 Spread OVN metadata agent heartbeat response in time\nca2e593177 Change grenade job names suffix to \"skip-level\"\n240f2c6aeb Split Hash Ring probing from the maintenance task\n1920f2a4e3 Add new oslo master CI jobs to the periodic queue\n7b8b797dea Load the required configuration options in the UT classes\n4858315286 Fix misplaced comparison constant warnings\n3d0f1ef4fa Remove some unnecessary list comprehensions\n70a2029ca1 [OVN] Set the default OVN metadata worker number to 0\n34b81ae9b7 Remove OVN gap minimum bandwidth support\n229d798a45 [OVN] Allow only one physical network per bridge\nf43891bf86 [OVN] Allow to execute ``MetadataProxyHandler`` in a local\nthread\n50eee19723 ovn: Use ovsdb-client to create neutron_pg_drop\n536498a29a Since OVN 20.06, config is stored in \"Chassis.other_config\"\n0c9e7d4e18 Update grenade skip level jobs for new release\nde18dd2c2e [ovn migration] Use ecsda ssh key instead of rsa\nd7b44f7218 Allow to pass EUI64 IP address as fixed ip for the port\na3e68e8f76 Fix ipam_pluggable_backend unit tests module\na8a43c8045 Execute \"IpMonitorTestCase\" tests always inside a namespace\nf8fa909444 Since OVN 20.06, config is stored in \"Chassis.other_config\"\n7b15ad9433 Bump minimum OVN version to v21.06.0\ne4cc40f114 fix: Fix url of Floodlight\n21491efd9f Port provisioning should retry only for VM ports\n178ee6fd3d [Trunk] Update the trunk status with the parent status\n0b2844b8a8 Add .pylintrc to irrelevant-files for non pep8 jobs\n72493d1472 Add missing comma in .pylintrc\n3f20cabc4f Revert \"Move the \"ovn_hash_ring\" clean up to maintenance\nworker\"\n56f89131dd docs: Remove errant indentation, mark up literals\n2f0919f7df Accept a port deletion with missing port binding information\n1fe4edab80 Open Antelope DB branch\n41a7479d0f Handle several dhcp agents for metadata over ipv6\n819a1bb3e6 Move the \"ovn_hash_ring\" clean up to maintenance worker\ne9b5a27001 Use \"OVNMechDriver\" instance in\n\"TestOvn[Nb|Sb]IdlNotifyHandler\"\n9eddb132b2 Fix openstack-tox-py39-with-oslo-master job\na765b7acbb [sqlalchemy-20] Add py39 oslo master to experimental\nea11928ba0 Neutron needs oslo.config \u003e\u003d 9\n9a1b5d1e51 Imported Translations from Zanata\n46bc1e7f13 Switch to 2023.1 Python3 unit tests and generic template name\n4914d7fab2 Update master for stable/zed\n639257c77b Make path for tcpdump release-specific\ncca48be73d Imported Translations from Zanata\nf7578e51f2 Collect ovn/ovs logs in singlenode jobs\n9655466763 [OVN] Rate limit the \"Disallow caching\" log from hash ring\n\nChange-Id: Iea7bef0e3e3915bea6859dabccdd032bba6717ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c1e7fefc44deb39a98d0e5c820c94dea23d88a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c1e7fefc44deb39a98d0e5c820c94dea23d88a4"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
