)]}'
{"id":"openstack%2Freleases~886987","triplet_id":"openstack%2Freleases~master~Ie32a4b008e8e64b423a2e2799488559eb4bd1ef7","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-06-27 12:46:46.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-06-28 14:35:01.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2023-06-28 14:35:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie32a4b008e8e64b423a2e2799488559eb4bd1ef7","subject":"release oslo.messaging 14.2.1","status":"MERGED","created":"2023-06-26 14:50:31.000000000","updated":"2023-06-28 14:35:01.000000000","submitted":"2023-06-28 14:35:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"886987","meta_rev_id":"618b2beeeeac729e0d3389592ba297b0512225bb","_number":886987,"virtual_id_number":886987,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-28 14:35:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-06-28 14:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-06-27 12:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-06-28 14:24:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-06-27 12:47:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-26 14:51:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-26 15:07:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-26 15:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-06-26 15:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2023-06-28 14:24:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"232269f007adf60439d3e976105313936dd5032a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-26 14:50:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6f5255c69e86c106a4177b57885029fe5783477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 15:07:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f599f95d72641128d2849f532a9db04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed94897ea714a829a521da799bd48e4 : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67e26664421447ecbfb9b631222ceee8 : FAILURE in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cde20c4914d24a74b7042959f0ac38a6 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"97edbb6489be8142c02b2fef255f8bcf25de3910","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-06-26 16:06:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a3db16c1107d4f53acd580471ca7b4c1d029643","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-27 08:32:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"af6a8d6c352290f456aed7aa3a125963114a3dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 08:49:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/276a6efb60eb4be080d9f120bd778fd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d04951580e4faa81054567aa755df7 : SUCCESS in 8m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/329168cd6b1d4b458cef7f4d6a877231 : SUCCESS in 6m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/431e80b0fb81410083ac4b6dae3d9cce : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"08528925726849fce2745505cb021b6b564db000","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-06-27 12:46:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"90bd5b694bf0a63727d8c5e2151884b66dfa46dd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 12:47:39.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9850d45819f670df228d1894dfc0afa740231884","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-06-28 14:24:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"67175caad5a1687ef323a9cfef457b3ea8ccc679","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:25:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"821216458b99ead299980e2ddafc64e92a2e2200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:35:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02670f59fa70484987733e81d8754893\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ad8aa172fe4d09875d5e0ef14ebb73 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf60d7695a84d239bbe2f836c064210 : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/763f8917730248d687675cc826c3e278 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"618b2beeeeac729e0d3389592ba297b0512225bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:35:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"63ae36a9fd6396f35f58f06b5f0c413fcfa0d0e4","revisions":{"f912d082ba3046b4e1fd4fa7d0527199779ff0a0":{"kind":"REWORK","_number":1,"created":"2023-06-26 14:50:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/87/886987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/886987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/886987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/886987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/886987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/886987/1"}}},"commit":{"parents":[{"commit":"fdb33212a9562367f587b56f604982f784a7fdfd","subject":"Merge \"[neutron-tempest-plugin] Tag xena-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fdb33212a9562367f587b56f604982f784a7fdfd"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-26 14:50:10.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-26 14:50:28.000000000","tz":120},"subject":"release oslo.messaging 14.2.1","message":"release oslo.messaging 14.2.1\n\nChange-Id: Ie32a4b008e8e64b423a2e2799488559eb4bd1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f912d082ba3046b4e1fd4fa7d0527199779ff0a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f912d082ba3046b4e1fd4fa7d0527199779ff0a0"}]},"branch":"refs/heads/master"},"63ae36a9fd6396f35f58f06b5f0c413fcfa0d0e4":{"kind":"REWORK","_number":2,"created":"2023-06-27 08:32:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/87/886987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/886987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/886987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/886987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/886987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/886987/2"}}},"commit":{"parents":[{"commit":"fdb33212a9562367f587b56f604982f784a7fdfd","subject":"Merge \"[neutron-tempest-plugin] Tag xena-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fdb33212a9562367f587b56f604982f784a7fdfd"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-26 14:50:10.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-27 08:32:31.000000000","tz":120},"subject":"release oslo.messaging 14.2.1","message":"release oslo.messaging 14.2.1\n\nChange-Id: Ie32a4b008e8e64b423a2e2799488559eb4bd1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63ae36a9fd6396f35f58f06b5f0c413fcfa0d0e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63ae36a9fd6396f35f58f06b5f0c413fcfa0d0e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
