)]}'
{"id":"openstack%2Freleases~889457","triplet_id":"openstack%2Freleases~master~I6a22f638513ff7f71a0f2ab2cc3e6cceeeff46c5","project":"openstack/releases","branch":"master","topic":"yoga-convenience-before-em","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-08-01 10:27:03.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-08-01 10:27:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a22f638513ff7f71a0f2ab2cc3e6cceeeff46c5","subject":"[Yoga] Release horizon","status":"MERGED","created":"2023-07-25 14:46:44.000000000","updated":"2023-08-01 10:27:03.000000000","submitted":"2023-08-01 10:27:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"889457-yoga-convenience-before-em","meta_rev_id":"8d7fc9ea2111345419376c616fb7c356e50dce7c","_number":889457,"virtual_id_number":889457,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-01 10:27:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2023-07-31 08:48:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2023-08-01 10:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-07-31 09:04:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2023-08-01 10:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-07-31 08:49:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 14:47:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-25 15:14:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-25 15:20:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2023-08-01 10:18:57.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"45a4b1a3e186d19ddeb158285ce4d6e89b3a548e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-25 14:46:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d3609e8ea1a54092f9230ff1fcc8d44993b17d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 15:14:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d58063baea2f4e56b01ff498eebd5a81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dad9e69885f4b0bb277f20871616288 : SUCCESS in 8m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98aef56bc9be4a9498af809a542125aa : FAILURE in 5m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f7e5aba8957240eb8bf10384c9b83113 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f577ee1ea66c6add4403fddd4628c3a55d67ebf8","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-26 03:22:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"14961735644607290a6b175ee80c48d8005e2a94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-28 08:56:17.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"14c47f23bfba7aaf1a06967ba6ba3bc1b9530a66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 09:32:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/190daab584794413b864183a4c674db4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ec21d6e8be04440bed2d64d45f44650 : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/434c2cf9391040578f20864f1aa6db92 : SUCCESS in 7m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0bd816a9ca8744cd8960827cf0be293d : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9cc469fa1e18e35e1e165b3f8eb51544e64c5ad0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-31 08:48:22.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"91af316a28be61438bb33b984a62a008a9fd57ed","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-31 08:48:51.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2afee370d5726c2ff40f7351d9ac67a4475fa042","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 08:49:13.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"4a5d34101bd0330d5a1203057a78df99c9d41211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 09:04:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d233d823f641e1a23e1d046fbcda11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0aedd9e3a642a5a4dcc0acb1052eec : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f48c2edab6674df18e01cadf3f1db05b : SUCCESS in 7m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b071c0ff99a84fe38f0897e531492a27 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"748c88df5c80d0d385dd43445b03dc3ea49af351","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-31 09:04:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"59500afc009ba2bc8006faa3e7a21e19bc19d3b1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-01 10:18:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"642219f224279e7dce677d8c6d4e1d4caad24942","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:19:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"946ed95f0a0937fe8c2c5f16d00c5a5efa7fdfdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:27:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40752407ab8a4974b5e6729df229819c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87114699df4548a6b101c740b060e57b : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea619315e6c64f578043405058a14dea : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5e35c3314c348479fac0cf13f73c4c9 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":3},{"id":"8d7fc9ea2111345419376c616fb7c356e50dce7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:27:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"04bc89840ad5e0729e95320c27bd5264b4aa84f8","revisions":{"83a6c1aed77ef50acdda0216df2d0ad13a3a7463":{"kind":"REWORK","_number":1,"created":"2023-07-25 14:46:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/57/889457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/889457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/889457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/889457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/889457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/889457/1"}}},"commit":{"parents":[{"commit":"3e4a235d313fd26629377ae553ae295298d7b066","subject":"Merge \"Release octavia 11.0.1 for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e4a235d313fd26629377ae553ae295298d7b066"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 14:46:38.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 14:46:38.000000000","tz":120},"subject":"[Yoga] Release horizon","message":"[Yoga] Release horizon\n\nThis is a convenience release before moving Yoga to extended maintenance\n(estimated 2023-11-02).\n\n$ git log --oneline --no-merges 22.1.0..c921f3b2a\nc921f3b2a Fix \"Set admin password\" checkbox visibility\nf14a54262 Remove usage of openstack/xstatic-font-awesome\na7702671d Disable form auto-complete on login form\nb8e1bc23a Remove Description from Edit Image Metadata\n447b6c724 Fix instance error with attribute cells.\n6bf934930 Fix delete image tests.\nb85e38051 Use SetUp and cleanup for the network pagination tests\nd8a415bb0 Make integration tests stop on first failure\nf20cc8faa Integration test navigation machinery for Angular pages\n730c987d9 Add setup and teardown to the volumes pagination tests\n3e29311ed Add setup and teardown to the instances pagination tests\n1ea855e31 Speed up integration tests\nf3a6f2bd8 Fix instance table behavior after deleting an instance\nb42ae4fda Pin tox\u003c4 for stable branches testing\n19d0db197 Fix flavor id auto generation\n6c5783cad Make readonly metadata properties unable to edit\n36ed6b03b Fix preselection for launch Instance dialog\n1cd150cdc Fix rescue of volume-based instances\n13e821a07 Fix app cred create without project_id for domain admins\nb2e0aff57 zuul: Declare queue at top level\n09dfd37ab Fix flavor specification at instance overview page\n618e44469 Don\u0027t try and update port security if its not changing\na9ae00b16 Make existing empty metadata properties optional\n\nChange-Id: I6a22f638513ff7f71a0f2ab2cc3e6cceeeff46c5\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83a6c1aed77ef50acdda0216df2d0ad13a3a7463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83a6c1aed77ef50acdda0216df2d0ad13a3a7463"}]},"branch":"refs/heads/master"},"5cc483e6f1578e1eededfef18ced8f1800f99863":{"kind":"REWORK","_number":2,"created":"2023-07-28 08:56:17.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/57/889457/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/889457/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/889457/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/889457/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/889457/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/889457/2"}}},"commit":{"parents":[{"commit":"3e4a235d313fd26629377ae553ae295298d7b066","subject":"Merge \"Release octavia 11.0.1 for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e4a235d313fd26629377ae553ae295298d7b066"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 14:46:38.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-07-28 08:56:17.000000000","tz":0},"subject":"[Yoga] Release horizon","message":"[Yoga] Release horizon\n\nThis is a convenience release before moving Yoga to extended maintenance\n(estimated 2023-11-02).\n\n$ git log --oneline --no-merges 22.1.0..c921f3b2a\nc921f3b2a Fix \"Set admin password\" checkbox visibility\nf14a54262 Remove usage of openstack/xstatic-font-awesome\na7702671d Disable form auto-complete on login form\nb8e1bc23a Remove Description from Edit Image Metadata\n447b6c724 Fix instance error with attribute cells.\n6bf934930 Fix delete image tests.\nb85e38051 Use SetUp and cleanup for the network pagination tests\nd8a415bb0 Make integration tests stop on first failure\nf20cc8faa Integration test navigation machinery for Angular pages\n730c987d9 Add setup and teardown to the volumes pagination tests\n3e29311ed Add setup and teardown to the instances pagination tests\n1ea855e31 Speed up integration tests\nf3a6f2bd8 Fix instance table behavior after deleting an instance\nb42ae4fda Pin tox\u003c4 for stable branches testing\n19d0db197 Fix flavor id auto generation\n6c5783cad Make readonly metadata properties unable to edit\n36ed6b03b Fix preselection for launch Instance dialog\n1cd150cdc Fix rescue of volume-based instances\n13e821a07 Fix app cred create without project_id for domain admins\nb2e0aff57 zuul: Declare queue at top level\n09dfd37ab Fix flavor specification at instance overview page\n618e44469 Don\u0027t try and update port security if its not changing\na9ae00b16 Make existing empty metadata properties optional\n\nChange-Id: I6a22f638513ff7f71a0f2ab2cc3e6cceeeff46c5\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5cc483e6f1578e1eededfef18ced8f1800f99863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5cc483e6f1578e1eededfef18ced8f1800f99863"}]},"branch":"refs/heads/master"},"04bc89840ad5e0729e95320c27bd5264b4aa84f8":{"kind":"REWORK","_number":3,"created":"2023-07-31 08:48:22.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/57/889457/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/889457/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/889457/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/889457/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/889457/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/889457/3"}}},"commit":{"parents":[{"commit":"3e4a235d313fd26629377ae553ae295298d7b066","subject":"Merge \"Release octavia 11.0.1 for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e4a235d313fd26629377ae553ae295298d7b066"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 14:46:38.000000000","tz":120},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2023-07-31 08:48:22.000000000","tz":0},"subject":"[Yoga] Release horizon","message":"[Yoga] Release horizon\n\nThis is a convenience release before moving Yoga to extended maintenance\n(estimated 2023-11-02).\n\n$ git log --oneline --no-merges 22.1.0..c921f3b2a\nc921f3b2a Fix \"Set admin password\" checkbox visibility\nf14a54262 Remove usage of openstack/xstatic-font-awesome\na7702671d Disable form auto-complete on login form\nb8e1bc23a Remove Description from Edit Image Metadata\n447b6c724 Fix instance error with attribute cells.\n6bf934930 Fix delete image tests.\nb85e38051 Use SetUp and cleanup for the network pagination tests\nd8a415bb0 Make integration tests stop on first failure\nf20cc8faa Integration test navigation machinery for Angular pages\n730c987d9 Add setup and teardown to the volumes pagination tests\n3e29311ed Add setup and teardown to the instances pagination tests\n1ea855e31 Speed up integration tests\nf3a6f2bd8 Fix instance table behavior after deleting an instance\nb42ae4fda Pin tox\u003c4 for stable branches testing\n19d0db197 Fix flavor id auto generation\n6c5783cad Make readonly metadata properties unable to edit\n36ed6b03b Fix preselection for launch Instance dialog\n1cd150cdc Fix rescue of volume-based instances\n13e821a07 Fix app cred create without project_id for domain admins\nb2e0aff57 zuul: Declare queue at top level\n09dfd37ab Fix flavor specification at instance overview page\n618e44469 Don\u0027t try and update port security if its not changing\na9ae00b16 Make existing empty metadata properties optional\n\nChange-Id: I6a22f638513ff7f71a0f2ab2cc3e6cceeeff46c5\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/04bc89840ad5e0729e95320c27bd5264b4aa84f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/04bc89840ad5e0729e95320c27bd5264b4aa84f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
