)]}'
{"id":"openstack%2Freleases~889498","triplet_id":"openstack%2Freleases~master~I6204808921f4fa8f253a2604fa1d47de9554b3c0","project":"openstack/releases","branch":"master","topic":"yoga-convenience-before-em","attention_set":{},"removed_from_attention_set":{"35988":{"account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"last_update":"2023-08-08 08:42:35.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-08-08 08:42:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6204808921f4fa8f253a2604fa1d47de9554b3c0","subject":"[Yoga] Release monasca-persister","status":"ABANDONED","created":"2023-07-25 15:14:28.000000000","updated":"2023-08-08 08:42:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b4d8281dba3f4e1229a2d391efe2ce188e308cf7","_number":889498,"virtual_id_number":889498,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-28 13:28:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 15:15:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-25 15:20:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"state":"REVIEWER"},{"updated":"2023-07-25 15:41:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f6b94e71ff086b9ca38c9a0b37662ab523c6178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-25 15:14:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9085b550484f42262a524087df21c9e5f467b9bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 15:41:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3f4094254934c13bbb3fa4e0f3acab1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa1b0ee7f514982ad4bb825432e166b : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e2915f30de244988df739f9492311e5 : FAILURE in 4m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4e1ceb88e3414a62aadfd2df7473b099 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d6ea2c4381f7e97023d923c2b0926534c3c35936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-28 13:19:39.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9501adb1ec76884deb19867ec1f40639a2f655c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 13:28:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b124f5362b8451b9c86f73c9b35f24c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67ba6653e3674f1fb0aa268d7d44db1b : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65c177911ce84c4fbf3f0002a7f18b3d : SUCCESS in 6m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1f570bec45c14e73b5c4c38cf2125ce2 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b4d8281dba3f4e1229a2d391efe2ce188e308cf7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-08 08:42:35.000000000","message":"Abandoned\n\nno functional changes, abandoning.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b44ae32c4c0abfe548e766625f484fbed6631598","revisions":{"80286d18f9c6134af68356004afcd14b053fa8f8":{"kind":"REWORK","_number":1,"created":"2023-07-25 15:14:28.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/98/889498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/889498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/889498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/889498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/889498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/889498/1"}}},"commit":{"parents":[{"commit":"3e4a235d313fd26629377ae553ae295298d7b066","subject":"Merge \"Release octavia 11.0.1 for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e4a235d313fd26629377ae553ae295298d7b066"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 15:14:20.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 15:14:20.000000000","tz":120},"subject":"[Yoga] Release monasca-persister","message":"[Yoga] Release monasca-persister\n\nThis is a convenience release before moving Yoga to extended maintenance\n(estimated 2023-11-02).\n\n$ git log --oneline --no-merges 6.0.0..21e6581\n21e6581 Drop lower-constraints.txt and its testing\nf04569c Update TOX_CONSTRAINTS_FILE for stable/yoga\nc8dfd12 Update .gitreview for stable/yoga\n\nChange-Id: I6204808921f4fa8f253a2604fa1d47de9554b3c0\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/80286d18f9c6134af68356004afcd14b053fa8f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/80286d18f9c6134af68356004afcd14b053fa8f8"}]},"branch":"refs/heads/master"},"b44ae32c4c0abfe548e766625f484fbed6631598":{"kind":"REWORK","_number":2,"created":"2023-07-28 13:19:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/98/889498/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/889498/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/889498/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/889498/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/889498/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/889498/2"}}},"commit":{"parents":[{"commit":"3e4a235d313fd26629377ae553ae295298d7b066","subject":"Merge \"Release octavia 11.0.1 for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e4a235d313fd26629377ae553ae295298d7b066"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 15:14:20.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-07-28 13:19:39.000000000","tz":0},"subject":"[Yoga] Release monasca-persister","message":"[Yoga] Release monasca-persister\n\nThis is a convenience release before moving Yoga to extended maintenance\n(estimated 2023-11-02).\n\n$ git log --oneline --no-merges 6.0.0..21e6581\n21e6581 Drop lower-constraints.txt and its testing\nf04569c Update TOX_CONSTRAINTS_FILE for stable/yoga\nc8dfd12 Update .gitreview for stable/yoga\n\nChange-Id: I6204808921f4fa8f253a2604fa1d47de9554b3c0\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b44ae32c4c0abfe548e766625f484fbed6631598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b44ae32c4c0abfe548e766625f484fbed6631598"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
