)]}'
{"id":"openstack%2Freleases~889576","triplet_id":"openstack%2Freleases~master~I3a8f1c1aa7e936bd307d7f59a98c34dd3d22c663","project":"openstack/releases","branch":"master","topic":"yoga-convenience-before-em","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-08-01 12:07:10.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-08-01 12:07:10.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-08-01 12:07:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a8f1c1aa7e936bd307d7f59a98c34dd3d22c663","subject":"[Yoga] Release puppet-tempest","status":"MERGED","created":"2023-07-25 16:10:42.000000000","updated":"2023-08-01 12:07:10.000000000","submitted":"2023-08-01 12:07:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889576-yoga-convenience-before-em","meta_rev_id":"7634f9359cb25ca4db2fbd09ad13f556851b7819","_number":889576,"virtual_id_number":889576,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-01 12:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-08-01 11:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-07-31 09:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-08-01 11:59:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-07-29 17:43:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 16:11:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-25 16:22:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-07-25 16:43:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-28 15:11:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-08-01 11:59:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4561012c921ad558c26848f007abeb7d77345665","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-25 16:10:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6f98d457fb291979387301dfbaf2f41e1ea4471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 16:43:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76d5cd7a0f06469ea09885d7cf3257ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe9433a73f2446fbafffe91a7c68734 : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffd75ed4e22f46dc9fc68db32beb500a : FAILURE in 6m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3fb4eaa68a1041fab6b2638a528f259d : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"94f5b9ec040125f60c67f457528d50027b1dbd0e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-28 09:35:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ef32e425bfe9f84cd93a1b9293ad90e7a6a2c34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-29 17:42:49.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0bf0923871dc3e23fef05892713e0fe854bf8e50","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-29 17:43:38.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"1e67dc6baeb02ce8f870338133f17be0e93efa59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-29 17:49:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2c17f1139c24f76b587d82ba4395450\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da8ad3daee44521a8fa7706577cff11 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41f91b5d796d45da984a274254d26b9c : SUCCESS in 4m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b16d0e09cb534a7f86647536d67832e2 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4d92169cfc26ef9ab2050243a9b3e94aa650ae21","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-31 09:03:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7afaaccaac8b2a5a4a435497fde562a9843e38cf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-01 11:59:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"175b56f860b476f9f469b53d83e97ebf4192b121","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 12:00:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"68111a5f5485fc452690a908e027d05499ef6f4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 12:07:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7bf24b3c96049aead4eaf95e7edf742\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49cda1a103bc442389c5c3f922e8fe6d : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/724ce3b74ce247a98572fbccb02dc2cd : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/760b140ea8134d2fab54fc4335dac0ba : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7634f9359cb25ca4db2fbd09ad13f556851b7819","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 12:07:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"adedb2a0b7fc50a6354f1159dd8a50e8aaec288b","revisions":{"c177c323b9996f5d2847b72d1e58d0ea08ff8778":{"kind":"REWORK","_number":1,"created":"2023-07-25 16:10:42.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/889576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/889576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/889576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/889576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/889576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/889576/1"}}},"commit":{"parents":[{"commit":"d462a3ce0a4357f868a14b96bc99be90e67d4235","subject":"Merge \"Add a Sorting function.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d462a3ce0a4357f868a14b96bc99be90e67d4235"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 16:10:36.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 16:10:36.000000000","tz":120},"subject":"[Yoga] Release puppet-tempest","message":"[Yoga] Release puppet-tempest\n\nThis is a convenience release before moving Yoga to extended maintenance\n(estimated 2023-11-02).\n\n$ git log --oneline --no-merges 20.3.0..dbc63ea\ne95a5fe Allow arbitrary section name for *_id_setter\n7feba89 Fix unit tests for plugin package management\n760bbba CentOS: Install tempest plugin packages only\n25af8e4 Change git url for Tempest to opendev.org\ncdd0ab8 Update TOX_CONSTRAINTS_FILE for stable/yoga\ncad8260 Update .gitreview for stable/yoga\n\nChange-Id: I3a8f1c1aa7e936bd307d7f59a98c34dd3d22c663\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c177c323b9996f5d2847b72d1e58d0ea08ff8778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c177c323b9996f5d2847b72d1e58d0ea08ff8778"}]},"branch":"refs/heads/master"},"adedb2a0b7fc50a6354f1159dd8a50e8aaec288b":{"kind":"REWORK","_number":2,"created":"2023-07-29 17:42:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/889576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/889576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/889576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/889576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/889576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/889576/2"}}},"commit":{"parents":[{"commit":"d462a3ce0a4357f868a14b96bc99be90e67d4235","subject":"Merge \"Add a Sorting function.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d462a3ce0a4357f868a14b96bc99be90e67d4235"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 16:10:36.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-29 17:42:48.000000000","tz":0},"subject":"[Yoga] Release puppet-tempest","message":"[Yoga] Release puppet-tempest\n\nThis is a convenience release before moving Yoga to extended maintenance\n(estimated 2023-11-02).\n\n$ git log --oneline --no-merges 20.3.0..dbc63ea\ne95a5fe Allow arbitrary section name for *_id_setter\n7feba89 Fix unit tests for plugin package management\n760bbba CentOS: Install tempest plugin packages only\n25af8e4 Change git url for Tempest to opendev.org\ncdd0ab8 Update TOX_CONSTRAINTS_FILE for stable/yoga\ncad8260 Update .gitreview for stable/yoga\n\nChange-Id: I3a8f1c1aa7e936bd307d7f59a98c34dd3d22c663\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/adedb2a0b7fc50a6354f1159dd8a50e8aaec288b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/adedb2a0b7fc50a6354f1159dd8a50e8aaec288b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
