)]}'
{"id":"openstack%2Freleases~889597","triplet_id":"openstack%2Freleases~master~Ied0a81d721ac953c69338b480845c525b968e4ad","project":"openstack/releases","branch":"master","topic":"yoga-convenience-before-em","attention_set":{},"removed_from_attention_set":{"35988":{"account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"last_update":"2023-08-07 15:57:56.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-08-07 15:57:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ied0a81d721ac953c69338b480845c525b968e4ad","subject":"[Yoga] Release python-monascaclient","status":"ABANDONED","created":"2023-07-25 16:18:34.000000000","updated":"2023-08-07 15:57:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"79e9c8cc1fc08980606fe5fe87681736a1371348","_number":889597,"virtual_id_number":889597,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-28 09:20:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 16:19:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-25 16:20:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"state":"REVIEWER"},{"updated":"2023-07-25 16:54:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"874e9c2d47ac03c057e697001702aa2cdaadda64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-25 16:18:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79562fa7cf6ad1d3c19eccc9a12c30aba3876fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 16:54:58.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c220996efcb54e7386e944ab6de1f9a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4276e7c58241422e85273b40c45c077c : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03c8679aebf7493389450f8fc0af6460 : FAILURE in 4m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a9b7d4306cca419ba45e113085c30c2a : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"10cf052927512c9e672f4cd7f8bb6a98d3941e7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-28 09:03:33.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ac453c1f2e2bb59b729c1365b1b4697532203861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 09:20:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28224bbc31804a118a7bf7f9a88bb2bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00210a903913426a9783af24e855b475 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fccf051f14534f5aaed7bd07f3420196 : SUCCESS in 5m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/13f8a857ba7341fcbd4b3249be20cca5 : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"79e9c8cc1fc08980606fe5fe87681736a1371348","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-07 15:57:56.000000000","message":"Abandoned\n\nNo functional change, hence is no worth of releasing this. Abandoning.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e170070a9fe01eacd3540149944e1ec66127684d","revisions":{"39c36b8ba3b945120466407900e25415775e46e2":{"kind":"REWORK","_number":1,"created":"2023-07-25 16:18:34.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/97/889597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/889597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/889597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/889597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/889597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/889597/1"}}},"commit":{"parents":[{"commit":"d462a3ce0a4357f868a14b96bc99be90e67d4235","subject":"Merge \"Add a Sorting function.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d462a3ce0a4357f868a14b96bc99be90e67d4235"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 16:18:29.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 16:18:29.000000000","tz":120},"subject":"[Yoga] Release python-monascaclient","message":"[Yoga] Release python-monascaclient\n\nThis is a convenience release before moving Yoga to extended maintenance\n(estimated 2023-11-02).\n\n$ git log --oneline --no-merges 2.5.0..40e2692\n40e2692 Drop lower-constraints.txt and its testing\n4e50acc Update TOX_CONSTRAINTS_FILE for stable/yoga\ncf9c182 Update .gitreview for stable/yoga\n\nChange-Id: Ied0a81d721ac953c69338b480845c525b968e4ad\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39c36b8ba3b945120466407900e25415775e46e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39c36b8ba3b945120466407900e25415775e46e2"}]},"branch":"refs/heads/master"},"e170070a9fe01eacd3540149944e1ec66127684d":{"kind":"REWORK","_number":2,"created":"2023-07-28 09:03:33.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/97/889597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/889597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/889597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/889597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/889597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/889597/2"}}},"commit":{"parents":[{"commit":"d462a3ce0a4357f868a14b96bc99be90e67d4235","subject":"Merge \"Add a Sorting function.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d462a3ce0a4357f868a14b96bc99be90e67d4235"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 16:18:29.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-07-28 09:03:32.000000000","tz":0},"subject":"[Yoga] Release python-monascaclient","message":"[Yoga] Release python-monascaclient\n\nThis is a convenience release before moving Yoga to extended maintenance\n(estimated 2023-11-02).\n\n$ git log --oneline --no-merges 2.5.0..40e2692\n40e2692 Drop lower-constraints.txt and its testing\n4e50acc Update TOX_CONSTRAINTS_FILE for stable/yoga\ncf9c182 Update .gitreview for stable/yoga\n\nChange-Id: Ied0a81d721ac953c69338b480845c525b968e4ad\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e170070a9fe01eacd3540149944e1ec66127684d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e170070a9fe01eacd3540149944e1ec66127684d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
