)]}'
{"id":"openstack%2Freleases~889618","triplet_id":"openstack%2Freleases~master~I4b89cc1a1b481eecddaba28522755742dfa0c36d","project":"openstack/releases","branch":"master","topic":"yoga-convenience-before-em","attention_set":{},"removed_from_attention_set":{"14107":{"account":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"last_update":"2023-07-26 13:04:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4b89cc1a1b481eecddaba28522755742dfa0c36d","subject":"[Yoga] Release solum-dashboard","status":"ABANDONED","created":"2023-07-25 16:28:45.000000000","updated":"2023-07-26 13:04:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"34277c8d1f58429b2b211a44e9f1ebe96de101e2","_number":889618,"virtual_id_number":889618,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-25 17:09:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 16:29:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-25 17:09:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-25 17:21:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"48f488aee55329a9bfb3fc89b0596febf5bc1561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-25 16:28:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01910cf904ebf0c1558d52de38f79b28e4aad266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 17:09:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a4968afc49d4c48b28aca9266d7d94e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/145558c99bc54c6ab720d37682f033c9 : SUCCESS in 8m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64039e1c390e47d2bb143d4d9ce71c16 : SUCCESS in 7m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ff15138d96784eb797d7e6f08d279dce : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"34277c8d1f58429b2b211a44e9f1ebe96de101e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-07-26 13:04:19.000000000","message":"Abandoned\n\nas discussed with Hervé on IRC, this release does not contain functional changes, so no need to release this. abandoning.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e92a70a3fd1c83f81b2ca3e9c565d8dbcded8666","revisions":{"e92a70a3fd1c83f81b2ca3e9c565d8dbcded8666":{"kind":"REWORK","_number":1,"created":"2023-07-25 16:28:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/18/889618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/889618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/889618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/889618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/889618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/889618/1"}}},"commit":{"parents":[{"commit":"d462a3ce0a4357f868a14b96bc99be90e67d4235","subject":"Merge \"Add a Sorting function.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d462a3ce0a4357f868a14b96bc99be90e67d4235"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 16:28:41.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-07-25 16:28:41.000000000","tz":120},"subject":"[Yoga] Release solum-dashboard","message":"[Yoga] Release solum-dashboard\n\nThis is a convenience release before moving Yoga to extended maintenance\n(estimated 2023-11-02).\n\n$ git log --oneline --no-merges 8.0.0..0e1623a\n0e1623a Update TOX_CONSTRAINTS_FILE for stable/yoga\nd00f932 Update .gitreview for stable/yoga\n\nChange-Id: I4b89cc1a1b481eecddaba28522755742dfa0c36d\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e92a70a3fd1c83f81b2ca3e9c565d8dbcded8666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e92a70a3fd1c83f81b2ca3e9c565d8dbcded8666"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
