)]}'
{"id":"openstack%2Freleases~892212","triplet_id":"openstack%2Freleases~master~Ib894e366e7916aa240b70c7462cf8783b2ab0cab","project":"openstack/releases","branch":"master","topic":"bobcat-final-non-client-libs","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-08-31 09:59:13.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-08-31 09:59:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib894e366e7916aa240b70c7462cf8783b2ab0cab","subject":"Release final openstacksdk for 2023.2 Bobcat","status":"MERGED","created":"2023-08-21 20:20:07.000000000","updated":"2023-08-31 09:59:13.000000000","submitted":"2023-08-31 09:59:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892212-bobcat-final-non-client-libs","meta_rev_id":"3731526a56cdd3b9427a7e86807a3c078b772168","_number":892212,"virtual_id_number":892212,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 09:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-08-31 09:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-08-30 14:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-08-31 09:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-08-31 09:46:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-08-29 10:51:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-21 20:20:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-21 20:35:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-21 21:22:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2023-08-31 09:07:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-31 09:46:11.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"93dfbd09cfa9589b66b682d09d409614f5c10794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-21 20:20:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3490d612163a8549c3a7e55463659b493955b333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 20:35:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa4d07420833428cabbc668506d67762\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50b4629b8c55426e92744d9d036c9fad : SUCCESS in 7m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cdd991d0e56845729c50ccf53acf60b6 : SUCCESS in 5m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2426419f72f54f329405bf022cf7d40c : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f3227738cd89a28f0bf190ede5efdc9c2d2090b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-08-24 15:15:01.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"832a698f3ad7c58b2f01af7f78507b608ba3175a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 15:15:48.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"67e7e464cff54df97a74c54a90173d85e124ea4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 15:28:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcdc740e750a433ebcf30102f0888541\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c64649541d4a01a79618b30e8b9ac2 : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c775eca6d654928947065954068c7ad : SUCCESS in 5m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e2bd414feff54237b78ef4111ee4d414 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"2c5f1574a67879d9c61e4c520c88ad4617cd2f2c","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-08-24 16:39:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8970e3f9088a3353dd6c611ba5146f87a2f42947","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-28 11:08:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e76b6ebe77395f23f0d88b49fa4e2df0667edbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-08-29 06:47:29.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d24f9a67646f37fe74d3a5696da74874e19f0e9e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 06:48:18.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"cd1110ede2543623fd168f98b218cdfdd0fa297e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 06:59:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c17f38f1ef714d7d9c7852a982b1b577\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba6a815f11349c9aef2ef10df90d515 : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c4e40f11cbe40eba7a3a6f768e671a1 : SUCCESS in 4m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/316bc269768f4c6596c134217d7b2978 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"8d0a5863c734ef18243210f20ce226e0c02a8528","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-08-29 10:12:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0a32eee562bd8920f164213f3da2b1be06164b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-08-29 10:51:04.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"842f3f49f5ed3464731fb94fa19aa91e9f9da419","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 10:51:52.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"ed350f335b225ef4293b3a3d72bf9d20c5cbadc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 11:03:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/721f4e824b6f46e380a4fca4763d8499\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd25462c5ea242b0a2478b49680a684c : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5466b3e10f84b89828bd52471671e0e : SUCCESS in 6m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e5984e9f723344b6bd76a7b5d0ea6b98 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":4},{"id":"e5b43487022d8b7cdbaf9a7cc6668bdbc2a7e02a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-29 11:12:06.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8588a19758be3d4592a301ba25c7e7c385a0517f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-08-29 13:14:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ccdde04f3582566954813afc04ffbeca5016033","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-30 14:40:03.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"248a25b3556dfc98e71c1797574df4548db191da","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-31 09:07:45.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c33c763a28761efcd3c618a637777623422d6fa8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-08-31 09:46:11.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"67318049be51e484c571250f3e337f42da0f317c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 09:46:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"57a5b7605936525a6603d013cdd32d528c2ba931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 09:59:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/075ed5fe3485472892c78c532196ab11\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be0040cb621245eebde2064128f3a901 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a80dd4024b4894b99c09d20a5870de : SUCCESS in 8m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0b2a9ba0cb046c3bc7905efd339835c : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":4},{"id":"3731526a56cdd3b9427a7e86807a3c078b772168","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 09:59:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"21f1b998e58052a1ad16e3ed8a725b5da88fa4b1","revisions":{"e70e0cd0359f2d6e40b9c91c5d92253311303683":{"kind":"REWORK","_number":1,"created":"2023-08-21 20:20:07.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/12/892212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/892212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/892212/1"}}},"commit":{"parents":[{"commit":"94e9bf64725d8a896cbdb0a43cb62b722930d179","subject":"Merge \"[solum-tempest-plugin] Tag xena-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94e9bf64725d8a896cbdb0a43cb62b722930d179"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-21 20:20:02.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-21 20:20:02.000000000","tz":120},"subject":"Release final openstacksdk for 2023.2 Bobcat","message":"Release final openstacksdk for 2023.2 Bobcat\n\nThis proposed a release for openstacksdk to get any final merged\nchanges out prior to the lib freeze for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, August 25th.\n\n$ git log --oneline --no-merges 1.4.0..21aade2f\nc05e17a0 image: Fixed URL for stores_info\nc6aaae56 Add image metadef object operations\n7b5356f3 Support manage/unmanage shares with manila\n8f9aa3b7 fix memory leak of Connections\nbb4f5f1e tests: Skip intermittently failing placement traits test\n8ee8f57a Stop randomly sorting error lines\n460cee1f Cleanup logic to either dry-run, bulk_delete or do single object deletes\nd7cb495b Fix manila access rules functional tests\n84f28505 tox: Add environment documentation\n82ee4444 cloud: Remove \u0027_object_store_client\u0027\n7852d957 cloud: Remove \u0027_{service}_client\u0027 properties\n0f7aeda9 cloud: Remove \u0027_is_client_version\u0027\n6ef2af39 Implemented methods for share metadata\n\nChange-Id: Ib894e366e7916aa240b70c7462cf8783b2ab0cab\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e70e0cd0359f2d6e40b9c91c5d92253311303683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e70e0cd0359f2d6e40b9c91c5d92253311303683"}]},"branch":"refs/heads/master"},"468675682f2818ffc027f4177b649c686e083c21":{"kind":"REWORK","_number":2,"created":"2023-08-24 15:15:01.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/12/892212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/892212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/892212/2"}}},"commit":{"parents":[{"commit":"94e9bf64725d8a896cbdb0a43cb62b722930d179","subject":"Merge \"[solum-tempest-plugin] Tag xena-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94e9bf64725d8a896cbdb0a43cb62b722930d179"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-21 20:20:02.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-08-24 15:15:01.000000000","tz":0},"subject":"Release final openstacksdk for 2023.2 Bobcat","message":"Release final openstacksdk for 2023.2 Bobcat\n\nThis proposed a release for openstacksdk to get any final merged\nchanges out prior to the lib freeze for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, August 25th.\n\n$ git log --oneline --no-merges 1.4.0..21aade2f\nc05e17a0 image: Fixed URL for stores_info\nc6aaae56 Add image metadef object operations\n7b5356f3 Support manage/unmanage shares with manila\n8f9aa3b7 fix memory leak of Connections\nbb4f5f1e tests: Skip intermittently failing placement traits test\n8ee8f57a Stop randomly sorting error lines\n460cee1f Cleanup logic to either dry-run, bulk_delete or do single object deletes\nd7cb495b Fix manila access rules functional tests\n84f28505 tox: Add environment documentation\n82ee4444 cloud: Remove \u0027_object_store_client\u0027\n7852d957 cloud: Remove \u0027_{service}_client\u0027 properties\n0f7aeda9 cloud: Remove \u0027_is_client_version\u0027\n6ef2af39 Implemented methods for share metadata\n\nChange-Id: Ib894e366e7916aa240b70c7462cf8783b2ab0cab\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/468675682f2818ffc027f4177b649c686e083c21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/468675682f2818ffc027f4177b649c686e083c21"}]},"branch":"refs/heads/master"},"c4ab1599b66807868033b4edd3a94cc30af9a463":{"kind":"REWORK","_number":3,"created":"2023-08-29 06:47:29.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/12/892212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/892212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/892212/3"}}},"commit":{"parents":[{"commit":"94e9bf64725d8a896cbdb0a43cb62b722930d179","subject":"Merge \"[solum-tempest-plugin] Tag xena-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94e9bf64725d8a896cbdb0a43cb62b722930d179"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-21 20:20:02.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-08-29 06:47:29.000000000","tz":0},"subject":"Release final openstacksdk for 2023.2 Bobcat","message":"Release final openstacksdk for 2023.2 Bobcat\n\nThis proposed a release for openstacksdk to get any final merged\nchanges out prior to the lib freeze for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, August 25th.\n\n$ git log --oneline --no-merges 1.4.0..21aade2f\nc05e17a0 image: Fixed URL for stores_info\nc6aaae56 Add image metadef object operations\n7b5356f3 Support manage/unmanage shares with manila\n8f9aa3b7 fix memory leak of Connections\nbb4f5f1e tests: Skip intermittently failing placement traits test\n8ee8f57a Stop randomly sorting error lines\n460cee1f Cleanup logic to either dry-run, bulk_delete or do single object deletes\nd7cb495b Fix manila access rules functional tests\n84f28505 tox: Add environment documentation\n82ee4444 cloud: Remove \u0027_object_store_client\u0027\n7852d957 cloud: Remove \u0027_{service}_client\u0027 properties\n0f7aeda9 cloud: Remove \u0027_is_client_version\u0027\n6ef2af39 Implemented methods for share metadata\n\nChange-Id: Ib894e366e7916aa240b70c7462cf8783b2ab0cab\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c4ab1599b66807868033b4edd3a94cc30af9a463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c4ab1599b66807868033b4edd3a94cc30af9a463"}]},"branch":"refs/heads/master"},"21f1b998e58052a1ad16e3ed8a725b5da88fa4b1":{"kind":"REWORK","_number":4,"created":"2023-08-29 10:51:04.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/12/892212/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/892212/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/892212/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/892212/4"}}},"commit":{"parents":[{"commit":"94e9bf64725d8a896cbdb0a43cb62b722930d179","subject":"Merge \"[solum-tempest-plugin] Tag xena-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94e9bf64725d8a896cbdb0a43cb62b722930d179"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-21 20:20:02.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-08-29 10:51:03.000000000","tz":0},"subject":"Release final openstacksdk for 2023.2 Bobcat","message":"Release final openstacksdk for 2023.2 Bobcat\n\nThis proposed a release for openstacksdk to get any final merged\nchanges out prior to the lib freeze for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, August 25th.\n\n$ git log --oneline --no-merges 1.4.0..21aade2f\nc05e17a0 image: Fixed URL for stores_info\nc6aaae56 Add image metadef object operations\n7b5356f3 Support manage/unmanage shares with manila\n8f9aa3b7 fix memory leak of Connections\nbb4f5f1e tests: Skip intermittently failing placement traits test\n8ee8f57a Stop randomly sorting error lines\n460cee1f Cleanup logic to either dry-run, bulk_delete or do single object deletes\nd7cb495b Fix manila access rules functional tests\n84f28505 tox: Add environment documentation\n82ee4444 cloud: Remove \u0027_object_store_client\u0027\n7852d957 cloud: Remove \u0027_{service}_client\u0027 properties\n0f7aeda9 cloud: Remove \u0027_is_client_version\u0027\n6ef2af39 Implemented methods for share metadata\n\nChange-Id: Ib894e366e7916aa240b70c7462cf8783b2ab0cab\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21f1b998e58052a1ad16e3ed8a725b5da88fa4b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21f1b998e58052a1ad16e3ed8a725b5da88fa4b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
