)]}'
{"id":"openstack%2Freleases~892241","triplet_id":"openstack%2Freleases~master~I47f2869de1bd238bed1deb13f7586b596a7b29cb","project":"openstack/releases","branch":"master","topic":"bobcat-final-non-client-libs","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-08-29 11:35:36.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2023-08-29 11:22:47.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-08-29 11:35:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I47f2869de1bd238bed1deb13f7586b596a7b29cb","subject":"Release final glance_store for 2023.2 Bobcat","status":"MERGED","created":"2023-08-21 20:54:52.000000000","updated":"2023-08-29 11:35:36.000000000","submitted":"2023-08-29 11:35:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892241-bobcat-final-non-client-libs","meta_rev_id":"d4daf51a4962bde6c8c83a56bef8a6d9fe541dc9","_number":892241,"virtual_id_number":892241,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 11:35:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-08-29 11:22:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-08-28 07:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-08-24 09:21:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-08-29 11:22:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-08-24 09:21:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-21 20:55:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-21 21:04:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-21 21:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2023-08-24 08:40:37.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-08-24 09:21:47.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"693dfd88cdd92c80ebbfbe0bd24a6963804e5fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-21 20:54:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78cf92d1236552428be01a3d5ecc7d6fcba5e1fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 21:04:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecce83e0f5bb4417be028dbadfd3c4fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e7c24514914a0b814c9736b30a2e3c : SUCCESS in 7m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97b3d15ea5cf427db4e88cf2dc3a14c6 : SUCCESS in 6m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/abe2356b88744680a0b6cf77cde31b05 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3ead46729fa22de5e3af26517fe90c3c3a28705a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 07:03:44.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d8110ed3bad6c8179d7ce73f22b64539a2bcb9d0","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 07:03:55.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bd753ac7b48e9d8b24fb682c78f67f586e832b3c","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 07:04:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"307eac17006b08586157bf6a7c7acd0878d687e4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 07:04:32.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"8068c590b2f8076b586c586f17103f1970f16d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 07:15:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14eadd80a5f1425f83d89c488658d4db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a87c201701949bea8b7ca25411b1e12 : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/231fe5502e8649cd9246f11d109bbf04 : SUCCESS in 5m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f5561e17a6654ddebabfd084b318c59d : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"6e16de4691e06ba30c63e3320dcf63c946094b05","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-08-24 08:40:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7aa3978899af15aa4408457f87823f0f927cbac8","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 09:20:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"53bde8e1cf3477f63b0f4c824fa24123b7e0b96c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 09:20:30.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3c52d3d7ef2e6523f0d8811e4f0ec48b47b54dc5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 09:21:24.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"e758b4519774103428563416e783237491fe423a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-24 09:21:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5136e33c9644b6409fe3fca5098a8fd06a727bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 09:27:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b3162d7f0c740aeb0f8e334beca9260\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46bf41112e10439497ef6ef781dcb279 : SUCCESS in 6m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec3a6ad45d9a44a983687b29645efcc6 : SUCCESS in 5m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bf8faf31241b40759b78b439e470d0da : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":3},{"id":"260d08de5ea8838d4dbba069590d6cc20a700496","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-28 07:53:50.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"139077dfb1cda730e57ee4f83f702b45bb09b280","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-08-29 11:22:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f3cceb56332c90fe9bf59d3167c8aa1285ba38b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 11:23:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2b2aaa97e34570086bd752c46578f4e8de829e75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 11:35:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e62bdfcbf314ab695a3a9659d549881\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b1a6afb123341fca76c9e57f51351f3 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6795d628f494a19abca50caabe1a1de : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b26620672c8f4124bd04252b06c46d58 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"d4daf51a4962bde6c8c83a56bef8a6d9fe541dc9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 11:35:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2c72b79c51ab58b4a3b0482490106afeb9b90370","revisions":{"e9d488a8eb7cb328affc54c74fb2e79d4db5a50e":{"kind":"REWORK","_number":1,"created":"2023-08-21 20:54:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/41/892241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/892241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/892241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/892241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/892241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/892241/1"}}},"commit":{"parents":[{"commit":"5a5da73ee07ca44a7535396ffe407b46a8cb6ed3","subject":"Merge \"Set Rocky series state as End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a5da73ee07ca44a7535396ffe407b46a8cb6ed3"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-08-21 20:51:34.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-08-21 20:51:34.000000000","tz":120},"subject":"Release final glance_store for 2023.2 Bobcat","message":"Release final glance_store for 2023.2 Bobcat\n\nThis proposed a release for glance_store to get any final merged\nchanges out prior to the lib freeze for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, August 25th.\n\n$ git log --oneline --no-merges 4.6.0..0c60291\n0c60291 Imported Translations from Zanata\n5a81f77 Imported Translations from Zanata\n\nChange-Id: I47f2869de1bd238bed1deb13f7586b596a7b29cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9d488a8eb7cb328affc54c74fb2e79d4db5a50e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9d488a8eb7cb328affc54c74fb2e79d4db5a50e"}]},"branch":"refs/heads/master"},"7e9dbf58d1781e670806550e317fb0a391d01280":{"kind":"REWORK","_number":2,"created":"2023-08-24 07:03:44.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/41/892241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/892241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/892241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/892241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/892241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/892241/2"}}},"commit":{"parents":[{"commit":"5a5da73ee07ca44a7535396ffe407b46a8cb6ed3","subject":"Merge \"Set Rocky series state as End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a5da73ee07ca44a7535396ffe407b46a8cb6ed3"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-08-21 20:51:34.000000000","tz":120},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-24 07:03:04.000000000","tz":0},"subject":"Release final glance_store for 2023.2 Bobcat","message":"Release final glance_store for 2023.2 Bobcat\n\nThis proposed a release for glance_store to get any final merged\nchanges out prior to the lib freeze for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, August 25th.\n\n$ git log --oneline --no-merges 4.6.0..0c60291\n0c60291 Imported Translations from Zanata\n5a81f77 Imported Translations from Zanata\n\nChange-Id: I47f2869de1bd238bed1deb13f7586b596a7b29cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e9dbf58d1781e670806550e317fb0a391d01280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e9dbf58d1781e670806550e317fb0a391d01280"}]},"branch":"refs/heads/master"},"2c72b79c51ab58b4a3b0482490106afeb9b90370":{"kind":"REWORK","_number":3,"created":"2023-08-24 09:20:30.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/41/892241/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/892241/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/892241/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/892241/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/892241/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/892241/3"}}},"commit":{"parents":[{"commit":"5a5da73ee07ca44a7535396ffe407b46a8cb6ed3","subject":"Merge \"Set Rocky series state as End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a5da73ee07ca44a7535396ffe407b46a8cb6ed3"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-08-21 20:51:34.000000000","tz":120},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-24 09:17:15.000000000","tz":0},"subject":"Release final glance_store for 2023.2 Bobcat","message":"Release final glance_store for 2023.2 Bobcat\n\nThis proposed a release for glance_store to get any final merged\nchanges out prior to the lib freeze for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, August 25th.\n\n$ git log --oneline --no-merges 4.6.0..0c60291\n0c60291 Imported Translations from Zanata\n5a81f77 Imported Translations from Zanata\n\nChange-Id: I47f2869de1bd238bed1deb13f7586b596a7b29cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c72b79c51ab58b4a3b0482490106afeb9b90370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c72b79c51ab58b4a3b0482490106afeb9b90370"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
