)]}'
{"id":"openstack%2Freleases~894717","triplet_id":"openstack%2Freleases~master~I22984b5e60e233002d8f5b799947e287bbba63a6","project":"openstack/releases","branch":"master","topic":"bobcat-rc1-deadline","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-09-20 08:57:18.000000000","reason":"Change was submitted"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2023-09-20 08:57:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I22984b5e60e233002d8f5b799947e287bbba63a6","subject":"Release tacker RC1 for 2023.2 Bobcat","status":"MERGED","created":"2023-09-12 11:36:31.000000000","updated":"2023-09-20 08:57:18.000000000","submitted":"2023-09-20 08:57:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894717-bobcat-rc1-deadline","meta_rev_id":"495d19776c7fc43a247fd891bc777cd6fb4e50d9","_number":894717,"virtual_id_number":894717,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-20 08:57:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":1,"date":"2023-09-15 15:43:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2023-09-15 15:58:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-09-20 08:47:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-09-20 08:47:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-09-15 15:27:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 11:37:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-12 11:43:46.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2023-09-12 11:48:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-20 08:47:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"d472ca9431d2d13f9cf86f370b279690dbea9239","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-12 11:36:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef9fb566570d1395ca882f0c9de19edf4b29227e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 11:48:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a9fe0372a3841a9a28eb07ed6534536\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b85a3ec8074f30ab5ae9171fb17183 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90b07773050845628c84248e6aca4215 : SUCCESS in 5m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b0f84631984a4dd08f332660c64385ae : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ce8ed15c7d97081e758d062a79ba008451fec13a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-09-15 02:02:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"09adc27812e0209ca99f132dc3fd575beebf6cb0","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-09-15 02:04:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"121764a0aa729908cfd67bd997ae5bf0e12dfbce","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 02:10:02.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"33b2e792b8c9a51d9b13a9dc0df3fb74352bd8f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 02:47:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c10b9df4b7134f08b73fdd5ed3517ab8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1c71d03c274932aef49bb8de0fe622 : SUCCESS in 8m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/444b75aba2b44bee92af4d7432f5b329 : SUCCESS in 7m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9c6de68585724b9bbf8a0a32ca21ebd8 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"66e0a1f7eb8f31dbb5a0b9803f55e926b25ea953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-09-15 14:08:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c049c0291508f1bcf9434a04b7f17633b1cae21d","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-09-15 14:09:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e141a3dd1f9f430639f6b9fa9b880b34c5f1738a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:20:09.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"8944dc6aeaa5e457115f262c8c5ed306fb560c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:32:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a00409fc0a97448482c9019ec8f5a162\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65055b1641884b8c9eeec44d40d1d252 : SUCCESS in 7m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c31edd3de5548e18fad73ae5538705b : SUCCESS in 5m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c9aa02f0692342d5b11bde0badec1fc0 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":3},{"id":"f37508eb7e2cb1fa6289f85b98c0ed78994227b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-09-15 15:27:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2318f73a07708d9a7e1adb01dcaf163e527ab8b6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 15:27:54.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"25e6b9b0092e99272b67ceb135426f2893560b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 15:39:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b05afc221064e85b308bce0cd26e6ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d4f42e4d45c457184c24db4e3d012cd : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/200109bb5e9346618060002e1258e809 : SUCCESS in 5m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/10efa9c2a4504843ae38fe57c447b725 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"db269e11b88fe18caea99359874846ae2b374e85","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-09-15 15:43:48.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f067fa1c08032cddc3bdc7213cbbd2dda3126858","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-15 15:58:44.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"64c23637bea53da2fb0248848410b90dbeb49ad0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-20 08:47:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bb4863d7d7b0dad1c81dea42f82c8a5c7b9c7668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 08:47:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"79c2eb161324b4656db8de96e70a02260bf1d3db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 08:57:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1b5f111859745f69d0bde2cf740950b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8070aeed1104066bd2adf8eaa599771 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c02858e36544ee29428a15b70e5075f : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4145ef069694c6989b1aaa0c9689fab : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":4},{"id":"495d19776c7fc43a247fd891bc777cd6fb4e50d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 08:57:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4098dfd0e837e5603f3552d1c983f47815c71dee","revisions":{"a1b4499ee227c4735eac6a82093786b5f45474cf":{"kind":"REWORK","_number":1,"created":"2023-09-12 11:36:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/17/894717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/894717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/894717/1"}}},"commit":{"parents":[{"commit":"a446036c395b3012ecdeca5f684ab108b2b3bfe8","subject":"Merge \"Release final python-barbicanclient for 2023.2 Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a446036c395b3012ecdeca5f684ab108b2b3bfe8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-12 11:36:27.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-12 11:36:27.000000000","tz":120},"subject":"Release tacker RC1 for 2023.2 Bobcat","message":"Release tacker RC1 for 2023.2 Bobcat\n\nThis proposed a RC1 release for tacker to get any merged changes\nout prior to the RC1 deadline for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 15.\n\nChange-Id: I22984b5e60e233002d8f5b799947e287bbba63a6\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1b4499ee227c4735eac6a82093786b5f45474cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1b4499ee227c4735eac6a82093786b5f45474cf"}]},"branch":"refs/heads/master"},"c036603a39ea9854ec1a146b88ca709bf74adab8":{"kind":"REWORK","_number":2,"created":"2023-09-15 02:02:29.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/17/894717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/894717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/894717/2"}}},"commit":{"parents":[{"commit":"43d95cc6584ab39f8c717e8d917093384febcb4a","subject":"Merge \"Release python-zaqarclient and cut stable/2023.2 branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/43d95cc6584ab39f8c717e8d917093384febcb4a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-12 11:36:27.000000000","tz":120},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2023-09-15 02:02:27.000000000","tz":540},"subject":"Release tacker RC1 for 2023.2 Bobcat","message":"Release tacker RC1 for 2023.2 Bobcat\n\nThis proposed a RC1 release for tacker to get any merged changes\nout prior to the RC1 deadline for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 15.\n\nChange-Id: I22984b5e60e233002d8f5b799947e287bbba63a6\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c036603a39ea9854ec1a146b88ca709bf74adab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c036603a39ea9854ec1a146b88ca709bf74adab8"}]},"branch":"refs/heads/master"},"b35c7d01cc8bbb9a1e6ff8bea11f13d57f1a3507":{"kind":"REWORK","_number":3,"created":"2023-09-15 14:08:07.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/17/894717/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/894717/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/894717/3"}}},"commit":{"parents":[{"commit":"43d95cc6584ab39f8c717e8d917093384febcb4a","subject":"Merge \"Release python-zaqarclient and cut stable/2023.2 branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/43d95cc6584ab39f8c717e8d917093384febcb4a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-12 11:36:27.000000000","tz":120},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2023-09-15 14:07:40.000000000","tz":540},"subject":"Release tacker RC1 for 2023.2 Bobcat","message":"Release tacker RC1 for 2023.2 Bobcat\n\nThis proposed a RC1 release for tacker to get any merged changes\nout prior to the RC1 deadline for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 15.\n\nChange-Id: I22984b5e60e233002d8f5b799947e287bbba63a6\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b35c7d01cc8bbb9a1e6ff8bea11f13d57f1a3507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b35c7d01cc8bbb9a1e6ff8bea11f13d57f1a3507"}]},"branch":"refs/heads/master"},"4098dfd0e837e5603f3552d1c983f47815c71dee":{"kind":"REWORK","_number":4,"created":"2023-09-15 15:27:05.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/17/894717/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/894717/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/894717/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/894717/4"}}},"commit":{"parents":[{"commit":"43d95cc6584ab39f8c717e8d917093384febcb4a","subject":"Merge \"Release python-zaqarclient and cut stable/2023.2 branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/43d95cc6584ab39f8c717e8d917093384febcb4a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-12 11:36:27.000000000","tz":120},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2023-09-15 15:26:53.000000000","tz":540},"subject":"Release tacker RC1 for 2023.2 Bobcat","message":"Release tacker RC1 for 2023.2 Bobcat\n\nThis proposed a RC1 release for tacker to get any merged changes\nout prior to the RC1 deadline for 2023.2 Bobcat going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 15.\n\nChange-Id: I22984b5e60e233002d8f5b799947e287bbba63a6\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4098dfd0e837e5603f3552d1c983f47815c71dee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4098dfd0e837e5603f3552d1c983f47815c71dee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
