)]}'
{"id":"openstack%2Freleases~895957","triplet_id":"openstack%2Freleases~master~I6f700ecd64bd6e102fe9ee0c48020c37a378aacb","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-09-20 17:03:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6f700ecd64bd6e102fe9ee0c48020c37a378aacb","subject":"metalsmith: Branch stable/2023.2 for bobcat","status":"ABANDONED","created":"2023-09-20 16:41:30.000000000","updated":"2023-09-20 17:03:26.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4143c20b38d027f9770633251da056a82ea53f4b","_number":895957,"virtual_id_number":895957,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-20 16:42:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-20 17:00:45.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"dc6bf378723cf79fefa74e2ab10f8c64bb816658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-20 16:41:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c693840fe273abd54fc26055a9fca2b05dd32055","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 16:42:17.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"cd4142f6016c0650cf4490300a391d68085c256e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 16:55:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad26af81e3234f51a8ac7ff3a573820d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9411b123a9b74119a427c89f41cc1e81 : SUCCESS in 8m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4dd57f5fb0b94f9e851c1319c335e5f6 : SUCCESS in 6m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a13a659dc2f345ff89a1b6d7cb153213 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8ae6bfdd21d45424d7029a27204a05ca2c86eb8a","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-09-20 17:00:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c79d50eae0590c0e424b7c4874525222c42579b8","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-09-20 17:02:17.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc3ecd8d96ff4376b12bfd9ef612666675e529cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-20 17:02:51.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0eac439d40e488d9b9d480513b0a75c3c9d6ae1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-20 17:03:10.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4143c20b38d027f9770633251da056a82ea53f4b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-20 17:03:26.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5ce5c311d898364162b5de4ffd6dd66c5c72489d","revisions":{"2186177a54ef5fe34d52d4f7f0c2564c1f547c20":{"kind":"REWORK","_number":1,"created":"2023-09-20 16:41:30.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/57/895957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/895957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/895957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/895957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/895957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/895957/1"}}},"commit":{"parents":[{"commit":"2efa61b9f4640bccc29acf2f26c9512706dbce5a","subject":"Merge \"Fix release id fetch for validate_series_open\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2efa61b9f4640bccc29acf2f26c9512706dbce5a"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-20 16:41:08.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-20 16:41:08.000000000","tz":-420},"subject":"metalsmith: 2.0.2 release for bobcat","message":"metalsmith: 2.0.2 release for bobcat\n\nBugfix release for bobcat.\n\nChange-Id: I6f700ecd64bd6e102fe9ee0c48020c37a378aacb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2186177a54ef5fe34d52d4f7f0c2564c1f547c20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2186177a54ef5fe34d52d4f7f0c2564c1f547c20"}]},"branch":"refs/heads/master"},"86be616fa44c7ae0ea776652a9bb6ce04aff3cd8":{"kind":"REWORK","_number":2,"created":"2023-09-20 17:02:51.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/57/895957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/895957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/895957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/895957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/895957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/895957/2"}}},"commit":{"parents":[{"commit":"2efa61b9f4640bccc29acf2f26c9512706dbce5a","subject":"Merge \"Fix release id fetch for validate_series_open\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2efa61b9f4640bccc29acf2f26c9512706dbce5a"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-20 16:41:08.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-20 17:02:51.000000000","tz":0},"subject":"metalsmith: 2.0.2 release for bobcat","message":"metalsmith: 2.0.2 release for bobcat\n\nBugfix release for bobcat.\n\nChange-Id: I6f700ecd64bd6e102fe9ee0c48020c37a378aacb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/86be616fa44c7ae0ea776652a9bb6ce04aff3cd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/86be616fa44c7ae0ea776652a9bb6ce04aff3cd8"}]},"branch":"refs/heads/master"},"5ce5c311d898364162b5de4ffd6dd66c5c72489d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-09-20 17:03:10.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/57/895957/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/895957/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/895957/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/895957/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/895957/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/895957/3"}}},"commit":{"parents":[{"commit":"2efa61b9f4640bccc29acf2f26c9512706dbce5a","subject":"Merge \"Fix release id fetch for validate_series_open\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2efa61b9f4640bccc29acf2f26c9512706dbce5a"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-20 16:41:08.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-20 17:03:10.000000000","tz":0},"subject":"metalsmith: Branch stable/2023.2 for bobcat","message":"metalsmith: Branch stable/2023.2 for bobcat\n\nBranching for bobcat.\n\nChange-Id: I6f700ecd64bd6e102fe9ee0c48020c37a378aacb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ce5c311d898364162b5de4ffd6dd66c5c72489d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ce5c311d898364162b5de4ffd6dd66c5c72489d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
