)]}'
{"id":"openstack%2Freleases~896784","triplet_id":"openstack%2Freleases~master~I93fbf2be0f0da85dedfe9b9f8d5876ca5b9e4b2e","project":"openstack/releases","branch":"master","topic":"fix-distributed-approval","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2023-11-30 09:01:20.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-11-30 09:15:34.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-11-30 09:15:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I93fbf2be0f0da85dedfe9b9f8d5876ca5b9e4b2e","subject":"Fix approval check for distributed leadership type","status":"MERGED","created":"2023-09-28 12:00:11.000000000","updated":"2023-11-30 09:15:34.000000000","submitted":"2023-11-30 09:15:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896784-fix-distributed-approval","meta_rev_id":"a7522b3c74665b0571d8f20f5b853e28df48af54","_number":896784,"virtual_id_number":896784,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-30 09:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-11-30 09:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-11-30 09:01:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-11-29 12:42:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 12:13:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 08:29:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-29 09:53:52.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2023-11-30 09:01:20.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"49663ed0f332999a1b46724c53f1d372f09172af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-28 12:00:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"767f330b61bd196f26d37ab629ff0cdeb7870a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 12:13:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cfd949173904a4780de727602d59d46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f4df875d9247108f507b93e47939d9 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc2a583e47204ed496a1c7057bffda34 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b70b9eea4504f5db009002ee60bc962 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca07e265d208453aabc7cfd59fbd8a78 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a380223838f4cf796f20d443e17bea8 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba9ac4ca52be4bd081b05115b543637e : SUCCESS in 7m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c08e871e298a4898b7305666a2744798 : SUCCESS in 4m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5b3b94319cbe4a96bceb1177092fe058 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5e59e5030cc998029cb2587af161ef452536d111","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-28 12:16:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"422adeb5cd4bfbc10128820912cb98b876fa3fd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 12:32:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8af9d5acdfe4a7d906ead9624da6833\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/344351c54db44747b43ddc3f08c21531 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/771f4319a9e34e1e954cec21d0053a08 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74386d1638054cf880503c7df7c980b5 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/574bd50df5524aa685bc11c3591299e4 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fea9626f62244c59a6bf6385a578c2d9 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c013f4a4b0c40128e45e6a7600002f4 : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07a458400fac4089851bc37b24c01aa5 : SUCCESS in 4m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f71b33e036244cc19e2b3a20e91fc8df : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9fb7b18d4d27ac334c41a45e48cfb2f8d466997d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-29 08:29:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5d3d7720ba204ea688130c48110658e7f995628a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-09-29 09:53:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22d02ed4c35e39dafaf9b1e3d2a7d689f49099f5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-29 10:10:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"308fa1f31aaf993ef822e5b1444007adcd6d32ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-28 18:18:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3baedd010ac229dd1ef780365ab3a6be57823c18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:29:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8c5c3deb7ca4d05bf4b658c89d9173b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61480e8d0b4c476b803416ce84000de7 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acf04fbc15ed4fad9ffe9e7c8e6431b2 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/100bb8a474a84cc39009b96d7fcd3741 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd277e586d094eab90e0815c7b339ca1 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b986a4e53f641acb8893dfd6d2dd26f : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a82202c7e778423e8d7433127b4f9668 : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd31a48b138a4f4a917a2477a07d7fd5 : SUCCESS in 4m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ff45727b32c44a09996722a59a55faba : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":3},{"id":"c91da84b37e6dce0513ebcfddce2d362584c4600","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-29 12:42:56.000000000","message":"Patch Set 3: PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb1db5235a09e0f406a6014a9fab99a377c1b7e4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-11-30 09:01:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4062dab2fea38834516410ae868fa067b9c039e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 09:01:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"13d3edcd4f7ce27b524d0440b8a8ed3504d1b473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 09:15:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c92a54cf00840b4b3bc09d1541fcc5c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6326ed9e7e3a457997a3ff8468789db2 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16125ae10f254230a8a61fc675152f4e : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4721d3d4506741e38cbb2d4a5a3f4457 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f94a5e264d0845f9b0c7575200c03644 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff0f902977794ab7aa7d9734cd14f44d : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f9187e733674fcc9d93b5f186574c15 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d909d7f432aa47eeaa73e94eb4be5d18 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"a7522b3c74665b0571d8f20f5b853e28df48af54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 09:15:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1f357b54e3cd7268dcd5fc54734eed900233b738","revisions":{"ef9f0ac7ebaa514dc939a1ea2709fa520000bf4f":{"kind":"REWORK","_number":1,"created":"2023-09-28 12:00:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/84/896784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/84/896784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/896784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/896784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/896784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/84/896784/1"}}},"commit":{"parents":[{"commit":"37c0fffe9f2d2f294b4a70430a14058dcaed11b3","subject":"Branch requirements for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/37c0fffe9f2d2f294b4a70430a14058dcaed11b3"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-09-28 11:43:30.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-09-28 11:50:02.000000000","tz":120},"subject":"Fix approval check for distributed leadership type","message":"Fix approval check for distributed leadership type\n\nTeams with distributed leadership type have release liaisons listed\nin governance repository\u0027s projects.yaml file instead of a single PTL.\nThis was introduced in the check_approval.py script, but it did not\nwork (though we did not realise this as usually releases repository\u0027s\nrelease_liaisons.yaml also contains the release liaisons). This patch\nfixes the check for distributed leadership type.\n\nChange-Id: I93fbf2be0f0da85dedfe9b9f8d5876ca5b9e4b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ef9f0ac7ebaa514dc939a1ea2709fa520000bf4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ef9f0ac7ebaa514dc939a1ea2709fa520000bf4f"}]},"branch":"refs/heads/master"},"7f3496bc41e1a0938b5d2376949e6faebb0d5400":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-28 12:16:08.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/84/896784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/84/896784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/896784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/896784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/896784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/84/896784/2"}}},"commit":{"parents":[{"commit":"9b14446832fdf49a802a8c55b7e5f6d664b12d4c","subject":"oslo.messaging 14.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b14446832fdf49a802a8c55b7e5f6d664b12d4c"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-09-28 11:43:30.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-09-28 12:15:41.000000000","tz":120},"subject":"Fix approval check for distributed leadership type","message":"Fix approval check for distributed leadership type\n\nTeams with distributed leadership type have release liaisons listed\nin governance repository\u0027s projects.yaml file instead of a single PTL.\nThis was introduced in the check_approval.py script, but it did not\nwork (though we did not realise this as usually releases repository\u0027s\nrelease_liaisons.yaml also contains the release liaisons). This patch\nfixes the check for distributed leadership type.\n\nChange-Id: I93fbf2be0f0da85dedfe9b9f8d5876ca5b9e4b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7f3496bc41e1a0938b5d2376949e6faebb0d5400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7f3496bc41e1a0938b5d2376949e6faebb0d5400"}]},"branch":"refs/heads/master"},"1f357b54e3cd7268dcd5fc54734eed900233b738":{"kind":"REWORK","_number":3,"created":"2023-11-28 18:18:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/84/896784/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/84/896784/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/896784/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/896784/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/896784/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/84/896784/3"}}},"commit":{"parents":[{"commit":"9b14446832fdf49a802a8c55b7e5f6d664b12d4c","subject":"oslo.messaging 14.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b14446832fdf49a802a8c55b7e5f6d664b12d4c"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-09-28 11:43:30.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-11-28 18:17:53.000000000","tz":60},"subject":"Fix approval check for distributed leadership type","message":"Fix approval check for distributed leadership type\n\nTeams with distributed leadership type have release liaisons listed\nin governance repository\u0027s projects.yaml file instead of a single PTL.\nThis was introduced in the check_approval.py script, but it did not\nwork (though we did not realise this as usually releases repository\u0027s\nrelease_liaisons.yaml also contains the release liaisons). This patch\nfixes the check for distributed leadership type.\n\nChange-Id: I93fbf2be0f0da85dedfe9b9f8d5876ca5b9e4b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1f357b54e3cd7268dcd5fc54734eed900233b738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1f357b54e3cd7268dcd5fc54734eed900233b738"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
