)]}'
{"id":"openstack%2Freleases~899602","triplet_id":"openstack%2Freleases~master~I226a9b0c655bad0cb74b6b21695a88d213ebd247","project":"openstack/releases","branch":"master","topic":"nova-bobcat-28.0.1","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-12-08 13:37:13.000000000","reason":"Change was submitted"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2023-12-08 13:37:13.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-11-30 13:39:29.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2023-12-08 13:37:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I226a9b0c655bad0cb74b6b21695a88d213ebd247","subject":"nova: Release 2023.2 Bobcat 28.0.1","status":"MERGED","created":"2023-10-30 15:51:04.000000000","updated":"2023-12-08 13:37:13.000000000","submitted":"2023-12-08 13:37:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"899602-nova-bobcat-28.0.1","meta_rev_id":"f713327541d5a8aadabe908deca847d61d8fc2ce","_number":899602,"virtual_id_number":899602,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-08 13:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-12-07 10:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-12-08 13:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-30 13:39:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-12-08 13:23:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-11-30 13:40:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-30 15:51:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-30 16:20:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-10-30 16:20:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2023-10-30 16:20:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2023-10-30 16:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 10:36:29.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-12-08 13:23:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"b5c0d5a0f04f8376347a583a34883b73a650dd5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-10-30 15:51:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2df265223b679e13de5c5b375d8ce866fa27fcc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 16:35:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9318bb028bb49498ecf9c0669edb798\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d7f7196aba42248401af876ea56ed8 : SUCCESS in 7m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e35f138d4664dad9a0b9895529d6155 : SUCCESS in 7m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a139b135a0874a96a0cad84320bbc827 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"68fc6a6e7d2e03e95ca1a766b87a4a5dd0ef3700","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-10-31 16:26:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26c9464742e28abc060f3504ff30a8aa3bdf1451","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-28 09:47:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"157173305b34a7c95393ddf9be05ea80a04b05ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 10:02:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ccb7190dc9d4fc5877ce86101bf17e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdcb027e10ea469f97b44d8d6ea9245d : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf6f5b37a2e14a38bb392300e7e2f67b : SUCCESS in 6m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7644664e452c4ad39968a2ca5d3090d5 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":2},{"id":"c999bf84cc4d1460161768ef1c52eb846e013e9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-30 10:17:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"205c864329c92931b33d39d9a94f2a37b7f5d31f","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-30 10:22:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"813c4113a884cae3c0d7bcec886b2779638f3766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 10:34:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da31c8bf36004bdfbf5a2422810d8bc7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f4fa75cfb5476ca6ba33ab147e0367 : SUCCESS in 7m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3efc32292424781a599cc723a914d59 : FAILURE in 8m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/efef242398dd462b86ea035dab804af6 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":3},{"id":"9144879fa36bac263999472a140238fb279e83f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-30 11:55:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7fe08c1306581a83c6168b31594910c4e2e47555","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 12:08:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8504b30781d14460bb00f678768d8c51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f13f00cf014d4fefb49f3fc0c36e4328 : SUCCESS in 6m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ca6214cef33453491ae2ffe6bcd984d : SUCCESS in 7m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/149b97a0ead04526a67ac89989a48767 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":4},{"id":"be17a5746aac15800621d184fd187804846ce028","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-30 13:39:29.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"39b76b1dce6caa6acf85162a699905141b568500","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 13:40:21.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"1af2d18e9bee83169b5283c88b0835fa23b10c20","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-12-07 10:36:29.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"68c6b26977cd4335f053282474aee0b24197de88","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-12-08 13:23:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"42f50c2a240fb73b933fb0c2f82b0bc39ea6bfb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 13:24:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"14e85568d463a8a78e21ab340690d39cc657d9b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 13:37:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad1a9cdfbac246e1b2c0894b461c6f0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a15e92a1a5824ae0b395f13a785d81a7 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51ddb06ea1e4e22b2c349a56d8f400f : SUCCESS in 6m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba013e446b134f60a3f341af88712ef4 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":4},{"id":"f713327541d5a8aadabe908deca847d61d8fc2ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 13:37:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a059dc7bd064ddaad08aa634587098b7237f99ba","revisions":{"ed65e79cc1a3d63f3c7765020d09fdd177af4b58":{"kind":"REWORK","_number":1,"created":"2023-10-30 15:51:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/899602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/899602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/899602/1"}}},"commit":{"parents":[{"commit":"d0ee7b1ecfe98fda578a1ca91df942b17e29117a","subject":"Merge \"Fix existing eol hashes on newton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0ee7b1ecfe98fda578a1ca91df942b17e29117a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-10-30 15:43:41.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-10-30 15:43:41.000000000","tz":60},"subject":"nova: Release 2023.2 Bobcat 28.0.1","message":"nova: Release 2023.2 Bobcat 28.0.1\n\n$ git log --oneline --no-merges 28.0.0..HEAD\na5e26bf6ca Revert \"Add upgrade check for compute-object-ids linkage\"\n6d5943e1b5 Install lxml before we need it in post-run\n7eeeb9d19d [stable-only] Update TOX_CONSTRAINTS_FILE for stable/2023.2\n6ac11b42b5 [stable-only] Update .gitreview for stable/2023.2\n\nChange-Id: I226a9b0c655bad0cb74b6b21695a88d213ebd247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ed65e79cc1a3d63f3c7765020d09fdd177af4b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ed65e79cc1a3d63f3c7765020d09fdd177af4b58"}]},"branch":"refs/heads/master"},"53aaf99577e2ddd8884548875b9a7c3ccba796c2":{"kind":"REWORK","_number":2,"created":"2023-11-28 09:47:23.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/899602/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/899602/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/899602/2"}}},"commit":{"parents":[{"commit":"d0ee7b1ecfe98fda578a1ca91df942b17e29117a","subject":"Merge \"Fix existing eol hashes on newton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0ee7b1ecfe98fda578a1ca91df942b17e29117a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-10-30 15:43:41.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-11-28 09:46:39.000000000","tz":60},"subject":"nova: Release 2023.2 Bobcat 28.0.1","message":"nova: Release 2023.2 Bobcat 28.0.1\n\n$ git log --oneline --no-merges 28.0.0..HEAD\n9cee58c6e4 Do not manage CPU0\u0027s state\n6b870ab90a Fix rebuild compute RPC API exception for rolling-upgrades\neb310f3bd2 add a regression test for all compute RPCAPI 6.x pinnings for rebuild\n7e4f45df91 Translate VF network capabilities to port binding\na5e26bf6ca Revert \"Add upgrade check for compute-object-ids linkage\"\n6d5943e1b5 Install lxml before we need it in post-run\n7eeeb9d19d [stable-only] Update TOX_CONSTRAINTS_FILE for stable/2023.2\n6ac11b42b5 [stable-only] Update .gitreview for stable/2023.2\n\nChange-Id: I226a9b0c655bad0cb74b6b21695a88d213ebd247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53aaf99577e2ddd8884548875b9a7c3ccba796c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53aaf99577e2ddd8884548875b9a7c3ccba796c2"}]},"branch":"refs/heads/master"},"13290e365833b5641f53cc8746c34403b60998e4":{"kind":"REWORK","_number":3,"created":"2023-11-30 10:17:19.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/899602/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/899602/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/899602/3"}}},"commit":{"parents":[{"commit":"d0ee7b1ecfe98fda578a1ca91df942b17e29117a","subject":"Merge \"Fix existing eol hashes on newton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0ee7b1ecfe98fda578a1ca91df942b17e29117a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-10-30 15:43:41.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-11-30 10:15:43.000000000","tz":60},"subject":"nova: Release 2023.2 Bobcat 28.0.1","message":"nova: Release 2023.2 Bobcat 28.0.1\n\n$ git log --oneline --no-merges 28.0.0..HEAD\n4549e34792 Allow enabling cpu_power_management with 0 dedicated CPUs\n9cee58c6e4 Do not manage CPU0\u0027s state\n6b870ab90a Fix rebuild compute RPC API exception for rolling-upgrades\neb310f3bd2 add a regression test for all compute RPCAPI 6.x pinnings for rebuild\n7e4f45df91 Translate VF network capabilities to port binding\na5e26bf6ca Revert \"Add upgrade check for compute-object-ids linkage\"\n6d5943e1b5 Install lxml before we need it in post-run\n7eeeb9d19d [stable-only] Update TOX_CONSTRAINTS_FILE for stable/2023.2\n6ac11b42b5 [stable-only] Update .gitreview for stable/2023.2\n\nChange-Id: I226a9b0c655bad0cb74b6b21695a88d213ebd247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13290e365833b5641f53cc8746c34403b60998e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13290e365833b5641f53cc8746c34403b60998e4"}]},"branch":"refs/heads/master"},"a059dc7bd064ddaad08aa634587098b7237f99ba":{"kind":"REWORK","_number":4,"created":"2023-11-30 11:55:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/899602/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/899602/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/899602/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/899602/4"}}},"commit":{"parents":[{"commit":"d0ee7b1ecfe98fda578a1ca91df942b17e29117a","subject":"Merge \"Fix existing eol hashes on newton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0ee7b1ecfe98fda578a1ca91df942b17e29117a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-10-30 15:43:41.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-11-30 11:55:08.000000000","tz":60},"subject":"nova: Release 2023.2 Bobcat 28.0.1","message":"nova: Release 2023.2 Bobcat 28.0.1\n\n$ git log --oneline --no-merges 28.0.0..HEAD\n4549e34792 Allow enabling cpu_power_management with 0 dedicated CPUs\n9cee58c6e4 Do not manage CPU0\u0027s state\n6b870ab90a Fix rebuild compute RPC API exception for rolling-upgrades\neb310f3bd2 add a regression test for all compute RPCAPI 6.x pinnings for rebuild\n7e4f45df91 Translate VF network capabilities to port binding\na5e26bf6ca Revert \"Add upgrade check for compute-object-ids linkage\"\n6d5943e1b5 Install lxml before we need it in post-run\n7eeeb9d19d [stable-only] Update TOX_CONSTRAINTS_FILE for stable/2023.2\n6ac11b42b5 [stable-only] Update .gitreview for stable/2023.2\n\nChange-Id: I226a9b0c655bad0cb74b6b21695a88d213ebd247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a059dc7bd064ddaad08aa634587098b7237f99ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a059dc7bd064ddaad08aa634587098b7237f99ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
