)]}'
{"id":"openstack%2Freleases~900041","triplet_id":"openstack%2Freleases~master~I43939be790d1c817deeb769d9116be5a769df3f1","project":"openstack/releases","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-11-03 16:53:11.000000000","reason":"Change was marked work in progress"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-11-07 18:27:35.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2023-11-07 18:27:35.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-11-07 18:27:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I43939be790d1c817deeb769d9116be5a769df3f1","subject":"Release PBR 6.0.0","status":"MERGED","created":"2023-11-03 13:19:47.000000000","updated":"2023-11-07 18:27:35.000000000","submitted":"2023-11-07 18:27:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900041-pip-23.1-support","meta_rev_id":"3782eedf4770921ad5b9e65c3e977c06365a2fb9","_number":900041,"virtual_id_number":900041,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-07 18:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-11-06 18:06:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-11-07 18:16:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-11-06 17:12:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-11-03 17:19:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2023-11-07 18:16:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-11-03 17:20:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-03 13:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-11-03 13:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2023-11-03 13:20:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-03 13:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 17:12:52.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-11-06 18:06:54.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-11-07 18:16:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"128b01781c397bebfa3a613d2eeade7b0b202e46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-03 13:19:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"455d61326ba191fb24f5a7030898505cb33be6b8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-11-03 13:22:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8ecf83959953839e872ab69ce03841b0fc647c85","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 13:23:49.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"5edba76652ff035937262023590d79dbc7184193","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 13:26:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fa80f8ef80149de81ec9305b8810c8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b3a553267e4759ade5feabe9f9d257 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/919ffe90095849f8aa8a3a211d20c829 : FAILURE in 5m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7cd83b8a765e4f89a13eaf04ea029327 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"be883d963f3a080acd2c78133912f72bc787da80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-03 15:54:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c15959f956647245012992c4d569a1c6a755b3db","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 15:55:39.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"04a5636685ef8f3a5c5d167fb03171fb355c4754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:05:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a77337da56134b62a2a4691356bb4f6c\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3cfd38e86e9e49d998dbb88ffc3e74b9 : FAILURE in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"12cc41936a7b98b35217a78d38c2f6b327fda7e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-11-03 16:22:09.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"944ba12b8dcc5511ee441dd8537571936058855f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:23:27.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"03aa35c4956fca09e11af0d5e8956d449a08d560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:28:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23e25ef16dbd42a3aa58adf60e6a2e01\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4defa29033342bc89d99a88c593d379 : FAILURE in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"f00e7e9fcfa6f17cdbb87035e0036222f3ef66f4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-03 16:53:11.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2f5c42a34615ff4918bd47b6be9479dc05ddaa13","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:54:36.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"00557087ab54275499c0f2c5c6bce86f4c479c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:59:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/334087df54054da8b8987df81b4d27be\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aca2730684d44e22be08dc72b0e7baa8 : FAILURE in 5m 20s","accounts_in_message":[],"_revision_number":4},{"id":"b9f1d533eff4cf59756ec5317bdc8d40ff9ba76b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-11-03 17:07:26.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a7bea57a2215bfba63275f03e3956e8768ddf659","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 17:08:44.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"ef7547d4307b677f4dd8730d039040d04405cbe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 17:15:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41355e9ebbb04a159d0c2834fdd3f775\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/555d50707c474e54be57ffd44c799ee1 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":5},{"id":"91baaef1f4bcee8bd2b4853bee551e70dfea9c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-03 17:19:52.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ce3cef0352c7f8f0717d0c4ed8e0566de3db2890","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 17:20:47.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"9da79983f1765791c8d60f59e983367e4fe3bab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 17:28:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80f9da8ed14643b796ed67833de87aad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c47ea1b28b824530adb3998fa9412662 : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bcac9a4669fb45a785530c1d34f8c789 : SUCCESS in 5m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c289511e280549558dcbf144c7dbe231 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":6},{"id":"405157edd64935546456b7f561fae56c7ebfe58a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-06 17:12:52.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"35136e9420f5e6849b7a17c2446c39fa6885574c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-11-06 18:06:54.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"690459571cb5ef6bd50c8b59e1a54b351999678e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-07 18:16:52.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"656da012d35a5156a4953dc92a668b3f5c8bb83b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 18:17:11.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a8a1203c72c8223f57b5ee1838db9a7c94b5d2a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 18:27:35.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bcaae8d465e4a42875cfedcaa634284\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/113e4b3f27eb4d37ab9fc0ce0c90d712 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72fcaf8ed2464d7abe1c54f8235838ae : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/414bc1ab7f1c452993167bf9477922b4 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":6},{"id":"3782eedf4770921ad5b9e65c3e977c06365a2fb9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 18:27:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4af8c3bfc3cfba50cf3a6a4ac885eef4e5992f4c","revisions":{"0b293d6e429eeadd75565ba61c43c87972336b3d":{"kind":"REWORK","_number":1,"created":"2023-11-03 13:19:47.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/41/900041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/900041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/900041/1"}}},"commit":{"parents":[{"commit":"f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412","subject":"Merge \"Move automaton to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 13:12:02.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 13:19:05.000000000","tz":0},"subject":"Release PBR 6.0.0","message":"Release PBR 6.0.0\n\nThis adds PEP 660 editable mode support, to address problems with\npyproject.toml build isolation when installing packages with a\nnon-default Python interpreter, whose urgency is increased by the\nneed to start doing PEP 517 in projects that rely on WSGI\nentrypoints now that pip 23.1 has started using its own fallback\npyproject.toml files.\n\nThis also removes support for Sphinx integration (broken with modern\nversions of Sphinx and SetupTools), thus the major version increase.\n\nChange-Id: I43939be790d1c817deeb769d9116be5a769df3f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b293d6e429eeadd75565ba61c43c87972336b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b293d6e429eeadd75565ba61c43c87972336b3d"}]},"branch":"refs/heads/master"},"882c59d10d7f68765c9653a1e5c94ccb2bb6b2b1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-03 15:54:14.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/41/900041/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/900041/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/900041/2"}}},"commit":{"parents":[{"commit":"5f70baf8835359cb820fea752c4b1137f1cd949d","subject":"DNM: Get more verbose logging for PBR release fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f70baf8835359cb820fea752c4b1137f1cd949d"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 13:12:02.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:54:08.000000000","tz":0},"subject":"Release PBR 6.0.0","message":"Release PBR 6.0.0\n\nThis adds PEP 660 editable mode support, to address problems with\npyproject.toml build isolation when installing packages with a\nnon-default Python interpreter, whose urgency is increased by the\nneed to start doing PEP 517 in projects that rely on WSGI\nentrypoints now that pip 23.1 has started using its own fallback\npyproject.toml files.\n\nThis also removes support for Sphinx integration (broken with modern\nversions of Sphinx and SetupTools), thus the major version increase.\n\nChange-Id: I43939be790d1c817deeb769d9116be5a769df3f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/882c59d10d7f68765c9653a1e5c94ccb2bb6b2b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/882c59d10d7f68765c9653a1e5c94ccb2bb6b2b1"}]},"branch":"refs/heads/master"},"23d462481fe04f1a3e5d814bb1264f4c953e9d75":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-03 16:22:09.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/41/900041/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/900041/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/900041/3"}}},"commit":{"parents":[{"commit":"25c9fcf70173e937c419339e5e72a53e7f1a815a","subject":"DNM: Get more verbose logging for PBR release fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25c9fcf70173e937c419339e5e72a53e7f1a815a"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 13:12:02.000000000","tz":0},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-11-03 16:22:02.000000000","tz":-420},"subject":"Release PBR 6.0.0","message":"Release PBR 6.0.0\n\nThis adds PEP 660 editable mode support, to address problems with\npyproject.toml build isolation when installing packages with a\nnon-default Python interpreter, whose urgency is increased by the\nneed to start doing PEP 517 in projects that rely on WSGI\nentrypoints now that pip 23.1 has started using its own fallback\npyproject.toml files.\n\nThis also removes support for Sphinx integration (broken with modern\nversions of Sphinx and SetupTools), thus the major version increase.\n\nChange-Id: I43939be790d1c817deeb769d9116be5a769df3f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23d462481fe04f1a3e5d814bb1264f4c953e9d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23d462481fe04f1a3e5d814bb1264f4c953e9d75"}]},"branch":"refs/heads/master"},"65f3a9c976dbbfb39d86a3de12e1ac3b7f4fa466":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-11-03 16:53:11.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/41/900041/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/900041/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/900041/4"}}},"commit":{"parents":[{"commit":"e8f6e3d587070ca7e2c125a50aeddcf0f9757b07","subject":"Fix PBR name check with new tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e8f6e3d587070ca7e2c125a50aeddcf0f9757b07"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 13:12:02.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 16:53:01.000000000","tz":0},"subject":"Release PBR 6.0.0","message":"Release PBR 6.0.0\n\nThis adds PEP 660 editable mode support, to address problems with\npyproject.toml build isolation when installing packages with a\nnon-default Python interpreter, whose urgency is increased by the\nneed to start doing PEP 517 in projects that rely on WSGI\nentrypoints now that pip 23.1 has started using its own fallback\npyproject.toml files.\n\nThis also removes support for Sphinx integration (broken with modern\nversions of Sphinx and SetupTools), thus the major version increase.\n\nChange-Id: I43939be790d1c817deeb769d9116be5a769df3f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/65f3a9c976dbbfb39d86a3de12e1ac3b7f4fa466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/65f3a9c976dbbfb39d86a3de12e1ac3b7f4fa466"}]},"branch":"refs/heads/master"},"95398485ec6e2576a1b9a8ddbc86c957f16b34f2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-11-03 17:07:26.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/41/900041/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/900041/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/900041/5"}}},"commit":{"parents":[{"commit":"9d0b6c85fc2ef8f2eea1c05efdacca78fd9daeec","subject":"Fix PBR name check with new tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d0b6c85fc2ef8f2eea1c05efdacca78fd9daeec"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 13:12:02.000000000","tz":0},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-11-03 17:07:07.000000000","tz":-420},"subject":"Release PBR 6.0.0","message":"Release PBR 6.0.0\n\nThis adds PEP 660 editable mode support, to address problems with\npyproject.toml build isolation when installing packages with a\nnon-default Python interpreter, whose urgency is increased by the\nneed to start doing PEP 517 in projects that rely on WSGI\nentrypoints now that pip 23.1 has started using its own fallback\npyproject.toml files.\n\nThis also removes support for Sphinx integration (broken with modern\nversions of Sphinx and SetupTools), thus the major version increase.\n\nChange-Id: I43939be790d1c817deeb769d9116be5a769df3f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95398485ec6e2576a1b9a8ddbc86c957f16b34f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95398485ec6e2576a1b9a8ddbc86c957f16b34f2"}]},"branch":"refs/heads/master"},"4af8c3bfc3cfba50cf3a6a4ac885eef4e5992f4c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-11-03 17:19:52.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/41/900041/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/900041/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/900041/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/900041/6"}}},"commit":{"parents":[{"commit":"cfb735ca383e71d7da293ee5ce0673920a3325bf","subject":"Fix PBR name check with new tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cfb735ca383e71d7da293ee5ce0673920a3325bf"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 13:12:02.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 17:19:14.000000000","tz":0},"subject":"Release PBR 6.0.0","message":"Release PBR 6.0.0\n\nThis adds PEP 660 editable mode support, to address problems with\npyproject.toml build isolation when installing packages with a\nnon-default Python interpreter, whose urgency is increased by the\nneed to start doing PEP 517 in projects that rely on WSGI\nentrypoints now that pip 23.1 has started using its own fallback\npyproject.toml files.\n\nThis also removes support for Sphinx integration (broken with modern\nversions of Sphinx and SetupTools), thus the major version increase.\n\nChange-Id: I43939be790d1c817deeb769d9116be5a769df3f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4af8c3bfc3cfba50cf3a6a4ac885eef4e5992f4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4af8c3bfc3cfba50cf3a6a4ac885eef4e5992f4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
