)]}'
{"id":"openstack%2Freleases~900053","triplet_id":"openstack%2Freleases~master~I35fb5dde9ec400397ed21035d57c3efb872d8e58","project":"openstack/releases","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-11-07 17:52:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35fb5dde9ec400397ed21035d57c3efb872d8e58","subject":"Fix PBR name check with new tox","status":"MERGED","created":"2023-11-03 15:22:09.000000000","updated":"2023-11-07 17:52:34.000000000","submitted":"2023-11-07 17:52:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900053-pip-23.1-support","meta_rev_id":"00e949c6cafd91005445202232ab8a880ce82a3f","_number":900053,"virtual_id_number":900053,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-07 17:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-11-06 19:16:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-11-06 17:13:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-11-06 12:37:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-11-07 17:38:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-11-07 17:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-03 15:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 12:37:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-11-06 17:13:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-11-06 19:16:48.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"26d888cf73d4fd77bf99fdf770bc17c00544f741","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-03 15:22:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d2e810f0cf460c44eae3ed0230ca4a5cdcb3eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 15:29:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d898d2028eb418d9587313eaebc7a7d\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ee6ff46d9fd4a66bae86b1bd247fbe1 : FAILURE in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f6746dc1896e6309a536cc77f2b7a2868d80b4d0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-03 15:36:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bc05002cb212694a92210dec5637464e45a8c15d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 15:51:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d763efcc86b453fb332b02ee59a8bc2\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af7b9b731b1a49129e38077836e0817e : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"cc4a115707dadda4c3c235c1bf9bbe507c8345de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-03 15:54:14.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"292e55c1fc4b0745f6224675c95796fbcc15f1ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:04:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d58f9d3f91e4a5c901f77fdc5fd1ba1\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70c9ad35ce1a4a33ae05a2831e27ee91 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":3},{"id":"779b1a4557f3d4556918680c58e3ae2e212e4c8b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-11-03 16:22:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"6fadfb4874f7c14443c8559b20ee4ed7ea1484a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:27:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9fb6fd325734e4a946e44cb34ba199d\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/677a7ee5744c47d3a3400bd07b517e76 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"c4869fdde20c7b625aa00ddd00346b7250afd470","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-03 16:53:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5c384234affaffed4e765e7b2f43151b2d815cd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 17:02:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46d5d2bf7bc14d5c857b38d2601ec443\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85da70985dbb4120ac29d61aea969c4d : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":5},{"id":"6f4b8e5c854adeb9ddc6816660a0bb63aaa1c97f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-11-03 17:07:26.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1b235bb193a8930a2c5860bc14924988e968a408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 17:13:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d8f1b19e3f248b5ba4252afb66001b1\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4bc64ce7b23f49f4ace91b66cbeb8153 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":6},{"id":"5882480b23dbd103df76d3f33812ac9a297d24bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-03 17:19:52.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2a78efc3dc329f9132d0a1440b831d9993a0dffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 17:42:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5be6e1e5c7e450a90b8533a0fbfda67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/224341e2cffe4cbcafaa1eb7a3c30dbb : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/151baa5dc1e945428f5f8592ebbd995d : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eaf28ec3a2fd4a8fb528c0bd5ef30028 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/220b75287fa3454caf4d86d34b9da03f : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa14dd87ab42406f9508fb251acd745e : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5272947bc3d471f8ef5a910a669e741 : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6edf50b97c8c4a6d91281b37f2a74bac : SUCCESS in 3m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ddc205c5e70d4c7d972671d05eb85975 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":7},{"id":"c869b04fb2a3eebae6063d00f2e62c38cf7cc677","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-11-06 12:37:36.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"9ecd5a63fa396a596b7f4353bef09555a5099e15","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-06 17:13:45.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"80cbce45118bac1d45640371bc3302e688aa91d5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-06 19:16:48.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"90802ed22d797faf886f7393424245233d58abef","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-07 17:38:54.000000000","message":"Patch Set 7: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"143105bfd289e8d020f103c0758284623c474af8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-07 17:39:21.000000000","message":"Patch Set 7: PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"638de48930d89b833cf908db8177a3c3abc2676e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 17:39:40.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"350d6353231881f073d667bb72fa50204755a0a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 17:52:34.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19e449fc89f240f58d34e8650473eb09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b9dda404f234d3c8338f448d968aaa5 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25f8c13c7d9148d3aee90ab718b29fbd : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8d36c169a1444389494289db08c5c94 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1552d9e6c324657ab858fa5845ab921 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89ed4f3820f04754922fb4ec8be19c3f : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12fa561b31f4a1c868dcd236cb63518 : SUCCESS in 6m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fde6f2d3d2946c3b6410f2ab508651b : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":7},{"id":"00e949c6cafd91005445202232ab8a880ce82a3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 17:52:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"cfb735ca383e71d7da293ee5ce0673920a3325bf","revisions":{"f9b16303c43d136ee4ed9ef98efa62fe93ea0969":{"kind":"REWORK","_number":1,"created":"2023-11-03 15:22:09.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/53/900053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/900053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/900053/1"}}},"commit":{"parents":[{"commit":"0b293d6e429eeadd75565ba61c43c87972336b3d","subject":"Release PBR 6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b293d6e429eeadd75565ba61c43c87972336b3d"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:19:58.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:21:55.000000000","tz":0},"subject":"DNM: Get more verbose logging for PBR release fail","message":"DNM: Get more verbose logging for PBR release fail\n\nChange-Id: I35fb5dde9ec400397ed21035d57c3efb872d8e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9b16303c43d136ee4ed9ef98efa62fe93ea0969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9b16303c43d136ee4ed9ef98efa62fe93ea0969"}]},"branch":"refs/heads/master"},"a5940ff45c83bbc1bc017e1ca8e68da783f94513":{"kind":"REWORK","_number":2,"created":"2023-11-03 15:36:52.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/53/900053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/900053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/900053/2"}}},"commit":{"parents":[{"commit":"0b293d6e429eeadd75565ba61c43c87972336b3d","subject":"Release PBR 6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b293d6e429eeadd75565ba61c43c87972336b3d"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:19:58.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:36:34.000000000","tz":0},"subject":"DNM: Get more verbose logging for PBR release fail","message":"DNM: Get more verbose logging for PBR release fail\n\nChange-Id: I35fb5dde9ec400397ed21035d57c3efb872d8e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5940ff45c83bbc1bc017e1ca8e68da783f94513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5940ff45c83bbc1bc017e1ca8e68da783f94513"}]},"branch":"refs/heads/master"},"5f70baf8835359cb820fea752c4b1137f1cd949d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-03 15:54:14.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/53/900053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/900053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/900053/3"}}},"commit":{"parents":[{"commit":"f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412","subject":"Merge \"Move automaton to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:19:58.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:54:08.000000000","tz":0},"subject":"DNM: Get more verbose logging for PBR release fail","message":"DNM: Get more verbose logging for PBR release fail\n\nChange-Id: I35fb5dde9ec400397ed21035d57c3efb872d8e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f70baf8835359cb820fea752c4b1137f1cd949d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f70baf8835359cb820fea752c4b1137f1cd949d"}]},"branch":"refs/heads/master"},"25c9fcf70173e937c419339e5e72a53e7f1a815a":{"kind":"REWORK","_number":4,"created":"2023-11-03 16:22:09.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/53/900053/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/900053/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/900053/4"}}},"commit":{"parents":[{"commit":"f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412","subject":"Merge \"Move automaton to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:19:58.000000000","tz":0},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-11-03 16:21:58.000000000","tz":-420},"subject":"DNM: Get more verbose logging for PBR release fail","message":"DNM: Get more verbose logging for PBR release fail\n\nChange-Id: I35fb5dde9ec400397ed21035d57c3efb872d8e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25c9fcf70173e937c419339e5e72a53e7f1a815a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25c9fcf70173e937c419339e5e72a53e7f1a815a"}]},"branch":"refs/heads/master"},"e8f6e3d587070ca7e2c125a50aeddcf0f9757b07":{"kind":"REWORK","_number":5,"created":"2023-11-03 16:53:11.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/53/900053/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/900053/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/900053/5"}}},"commit":{"parents":[{"commit":"f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412","subject":"Merge \"Move automaton to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:19:58.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 16:49:35.000000000","tz":0},"subject":"Fix PBR name check with new tox","message":"Fix PBR name check with new tox\n\nChanges to tox in the past year have rendered the PBR-specific\ntweaks in get_sdist_name invalid, since calling tox from within tox\nnow creates testenv directories relative to the parent tox\u0027s working\ndirectory rather than the child\u0027s. While we\u0027re here, turn on more\nverbose logging in the tox invocation, and fix the subprocess\nroutine to not discard the command\u0027s output.\n\nChange-Id: I35fb5dde9ec400397ed21035d57c3efb872d8e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e8f6e3d587070ca7e2c125a50aeddcf0f9757b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e8f6e3d587070ca7e2c125a50aeddcf0f9757b07"}]},"branch":"refs/heads/master"},"9d0b6c85fc2ef8f2eea1c05efdacca78fd9daeec":{"kind":"REWORK","_number":6,"created":"2023-11-03 17:07:26.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/53/900053/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/900053/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/900053/6"}}},"commit":{"parents":[{"commit":"f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412","subject":"Merge \"Move automaton to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:19:58.000000000","tz":0},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-11-03 17:06:55.000000000","tz":-420},"subject":"Fix PBR name check with new tox","message":"Fix PBR name check with new tox\n\nChanges to tox in the past year have rendered the PBR-specific\ntweaks in get_sdist_name invalid, since calling tox from within tox\nnow creates testenv directories relative to the parent tox\u0027s working\ndirectory rather than the child\u0027s. While we\u0027re here, turn on more\nverbose logging in the tox invocation, and fix the subprocess\nroutine to not discard the command\u0027s output.\n\nChange-Id: I35fb5dde9ec400397ed21035d57c3efb872d8e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d0b6c85fc2ef8f2eea1c05efdacca78fd9daeec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d0b6c85fc2ef8f2eea1c05efdacca78fd9daeec"}]},"branch":"refs/heads/master"},"cfb735ca383e71d7da293ee5ce0673920a3325bf":{"kind":"REWORK","_number":7,"created":"2023-11-03 17:19:52.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/53/900053/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/900053/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/900053/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/900053/7"}}},"commit":{"parents":[{"commit":"f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412","subject":"Merge \"Move automaton to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3298178a2e6f5bebe0ff9c0c3c25fe47aaa0412"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 15:19:58.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-11-03 17:19:12.000000000","tz":0},"subject":"Fix PBR name check with new tox","message":"Fix PBR name check with new tox\n\nChanges to tox in the past year have rendered the PBR-specific\ntweaks in get_sdist_name invalid, since calling tox from within tox\nnow creates testenv directories relative to the parent tox\u0027s working\ndirectory rather than the child\u0027s. While we\u0027re here, turn on more\nverbose logging in the tox invocation, and fix the subprocess\nroutine to not discard the command\u0027s output.\n\nChange-Id: I35fb5dde9ec400397ed21035d57c3efb872d8e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cfb735ca383e71d7da293ee5ce0673920a3325bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cfb735ca383e71d7da293ee5ce0673920a3325bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
