)]}'
{"id":"openstack%2Freleases~903635","triplet_id":"openstack%2Freleases~master~I98ac00e2ab3346eaeb02473b52b3e4b8ef00c8ed","project":"openstack/releases","branch":"master","topic":"cycle_names","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-04-30 18:13:31.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-12-20 08:50:17.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I98ac00e2ab3346eaeb02473b52b3e4b8ef00c8ed","subject":"[docs] Add symlink when cycle.release-ids differs from cycle.series_name","status":"ABANDONED","created":"2023-12-14 03:55:50.000000000","updated":"2024-04-30 18:13:31.000000000","total_comment_count":6,"unresolved_comment_count":2,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"16ab87e1bb0d016ca8806205ca9d4edf66d75c12","_number":903635,"virtual_id_number":903635,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-15 03:31:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2023-12-20 08:50:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-12-14 04:02:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-14 08:32:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-12-20 08:50:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"55a52a9fb02aad069bd1ebc3a2fda6889e2c9960","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-12-14 03:55:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e663f135500a7c3936f5411e0ad49e01ad052c28","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-12-14 03:57:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"650a91892719447281f75eac8e352aca37de71d3","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-12-14 03:57:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c42e28b4d90561a7167b3eab3a11ef0881b07de0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 04:02:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b2eab78d9b84287a617e8df6edcac1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf822af61004780bed50f253d32ef2f : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"06cd1b0f499e5a6a1edd3e8ce7d59c587761e6d4","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-12-14 06:01:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f58626e957c12237069f7a97148e205939ac1f20","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-14 08:32:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5728bd8d9775f00f54b58c7261d31acd703f892b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-12-15 01:47:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9caadfc63b8c5b187e7f7869e0408b123141298f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-12-15 03:14:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"591b4d42605187623dd4a719eb22ddf2c50cf40a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 03:31:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4e3ffd2402741138e7280c5e48b34a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0db794156074863a04d74b43e96eafb : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3db9e3ebb2d12ef4227e76c7ea2e41d391184836","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-20 08:50:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"16ab87e1bb0d016ca8806205ca9d4edf66d75c12","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-04-30 18:13:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a62054e0cba93b5d5f9f4fbd3d7de854937875f3","revisions":{"6ad46dff5b9811ba6ecb2234cdd7ae9626a0e053":{"kind":"REWORK","_number":1,"created":"2023-12-14 03:55:50.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/903635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/903635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/903635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/903635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/903635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/903635/1"}}},"commit":{"parents":[{"commit":"1439c9b343ada8c7abafd247a5bfb262744b6d22","subject":"Merge \"Release openstack-ansible for 2023.2 Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1439c9b343ada8c7abafd247a5bfb262744b6d22"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-12-14 03:51:20.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-12-14 03:51:20.000000000","tz":660},"subject":"[docs] Add symlink when cycle.release-ids differns from cycle.series_name","message":"[docs] Add symlink when cycle.release-ids differns from cycle.series_name\n\nThis is a \"flexible\" solution to make linking to specific cycle details\nless error prone.\n\nChange-Id: I98ac00e2ab3346eaeb02473b52b3e4b8ef00c8ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ad46dff5b9811ba6ecb2234cdd7ae9626a0e053"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ad46dff5b9811ba6ecb2234cdd7ae9626a0e053"}]},"branch":"refs/heads/master"},"a62054e0cba93b5d5f9f4fbd3d7de854937875f3":{"kind":"REWORK","_number":2,"created":"2023-12-15 03:14:16.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/903635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/903635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/903635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/903635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/903635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/903635/2"}}},"commit":{"parents":[{"commit":"1439c9b343ada8c7abafd247a5bfb262744b6d22","subject":"Merge \"Release openstack-ansible for 2023.2 Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1439c9b343ada8c7abafd247a5bfb262744b6d22"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-12-14 03:51:20.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-12-15 03:09:31.000000000","tz":660},"subject":"[docs] Add symlink when cycle.release-ids differs from cycle.series_name","message":"[docs] Add symlink when cycle.release-ids differs from cycle.series_name\n\nThis is a \"flexible\" solution to make linking to specific cycle details\nless error prone.  We\u0027ll need to follow-up with a change to insert\n\n`\u003clink rel\u003d\"canonical\"\n   href\u003d\"https://releases.openstack.org/${release_id}/${page}.html\" /\u003e`\n\ninto impacted pages.\n\nChange-Id: I98ac00e2ab3346eaeb02473b52b3e4b8ef00c8ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a62054e0cba93b5d5f9f4fbd3d7de854937875f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a62054e0cba93b5d5f9f4fbd3d7de854937875f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
