)]}'
{"id":"openstack%2Freleases~904750","triplet_id":"openstack%2Freleases~master~I59e647ff3a1ac33190feec982a5cea76cd372122","project":"openstack/releases","branch":"master","topic":"relmgt-unmaintained-state","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2024-01-04 15:32:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I59e647ff3a1ac33190feec982a5cea76cd372122","subject":"[test] release and cut stable branch for release-test","status":"MERGED","created":"2024-01-04 13:33:43.000000000","updated":"2024-01-04 15:32:42.000000000","submitted":"2024-01-04 15:32:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904750-relmgt-unmaintained-state","meta_rev_id":"bb6a4a53dc306a33f6ccf632b8b6a3868d57a663","_number":904750,"virtual_id_number":904750,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-04 15:32:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-01-04 15:23:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-01-04 15:23:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-01-04 13:40:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 13:40:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-04 14:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"01f135ae9a46b8dd54907c96bd1558e9f72adb4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-04 13:33:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9d6ffabb15253d39f92bc3c03e84b2c8926bca9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 13:40:31.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"1dfe1b14ec99920fc72d28279cd985d8a8966e6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 13:52:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5644156b0d954893bad42f0702d851b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dcbeb52836d44e888b58e3fc6a34496 : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69cf711b6ec341dca8ea309042f056fe : SUCCESS in 5m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c4a3a66743734deb84922a5b3a8428dc : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8ab00c86d5edebcbd25c99547786fdc77569a22d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-04 15:23:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e00c43a61e1b9a9db6c12ad014c45001907e15d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 15:24:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55ab10c8ad1ffaf1369c27ca55a00c7480652c47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 15:32:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70abac6f7d8247cf9db58eebb86fe32a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d70e6fd033a418b9ca3c4775b2a9578 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42db373d8ccd4a5d8716affcc07c3988 : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a51318b3c96d4a4c9c66228f5281d467 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bb6a4a53dc306a33f6ccf632b8b6a3868d57a663","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 15:32:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77e8ba577c0a83275d06b6f64c72cdaf1198d05d","revisions":{"77e8ba577c0a83275d06b6f64c72cdaf1198d05d":{"kind":"REWORK","_number":1,"created":"2024-01-04 13:33:43.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/50/904750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/904750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/904750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/904750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/904750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/904750/1"}}},"commit":{"parents":[{"commit":"cc7802a4f5104f02554010ac39cbeabbb8c1ff4c","subject":"Merge \"[glance] Transition Train to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cc7802a4f5104f02554010ac39cbeabbb8c1ff4c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-04 13:30:19.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-04 13:30:19.000000000","tz":60},"subject":"[test] release and cut stable branch for release-test","message":"[test] release and cut stable branch for release-test\n\nRelease and cut stable branch for release-test in Zed and Yoga series.\nSo that we can do tests with new Unmaintained changes on these\nbranches against release-test repository.\n\nChange-Id: I59e647ff3a1ac33190feec982a5cea76cd372122\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77e8ba577c0a83275d06b6f64c72cdaf1198d05d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77e8ba577c0a83275d06b6f64c72cdaf1198d05d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
