)]}'
{"id":"openstack%2Freleases~904767","triplet_id":"openstack%2Freleases~master~I606ba5e5f8fd783d6c3f83356ded27a46e8bac56","project":"openstack/releases","branch":"master","topic":"relmgt-unmaintained-state","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-01-15 13:21:52.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2024-01-15 12:54:40.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-01-15 13:21:52.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-01-12 14:46:27.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I606ba5e5f8fd783d6c3f83356ded27a46e8bac56","subject":"Introduce EOM tag validation","status":"MERGED","created":"2024-01-04 15:57:02.000000000","updated":"2024-01-15 13:21:52.000000000","submitted":"2024-01-15 13:21:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"904767-relmgt-unmaintained-state","meta_rev_id":"07ea9c843278cf80b8caa09fff0f1fce461b3e51","_number":904767,"virtual_id_number":904767,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-15 13:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2024-01-15 12:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-01-12 14:46:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2024-01-15 12:54:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-01-12 14:46:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2024-01-15 12:54:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 16:09:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-05 14:35:24.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2024-01-10 08:57:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-01-10 09:51:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1561f6739475b41d4415f0e48499b076b83110d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-04 15:57:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b020f45adae2ff3b9c873e9c6db87eaa9f9037b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 16:09:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57f83a65ee9f424b801e4ecf7af05de4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52fe8e46b35b4c2e8dd0594c840f518b : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e5393ecd08c4581ad76a9b13d2a3718 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70527a3c0e914526bde94d9c3d0df950 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9ccfdff37224c3c93fa177db6d91fff : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b75972aba4b402b945ad99d6eae664e : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59c191757ed540c1a8fe58b309593258 : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/495c10b3831d4f869164c33aeae90e2b : SUCCESS in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/890433bf103e4b3fb0199e03dbcf147a : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d5dca6eaaafcc77edca860a110a61127ae2c7c2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-04 16:22:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ba4b546a42b9347738e627ff88411a7ff23b140a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 16:36:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b7fc1d18f6f4821af24694ad1ae1366\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1a5380c445484ea4f3dcc015482693 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43863469ad5047e7ab0d36423e56bfb0 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25752ac0dddb426a82656de4335b1304 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1de9449cbe7414f9b86b842250f3353 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51e1adae8881446793e9bb8c90e04a30 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65762fea012443939f40413adbcc3da7 : SUCCESS in 7m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aaba9d09ecd14f1c8cb60bf65472117d : SUCCESS in 5m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c0606894755b4d5791f28ea056f3ea39 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"01d4c20178e2a2e082907fb265ce2828105b22aa","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-01-05 14:35:24.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f18bf8b0fd2a84d74cadba7ba8df2270878ffb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-10 08:53:29.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f8821d5ae8d8bf2c7a97e12d87e90acc988d08a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 09:27:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2345ec34b370479e9fd4d11f4c186533\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d575c2f5e4a54846823801dc2cc08163 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80785601aedd458c8b0b964ac68a013b : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6975d1cf6e3948d18bf1ca2415b7b438 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b365d063a4da400dac159b64f30a91f6 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28b251c0e5c14577aef8f02bca93b8f7 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a396c7658cd24e44bc5d1935f2d2fb56 : SUCCESS in 7m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4842b6b3ccad4f57ae417fd76cc818f8 : SUCCESS in 6m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ce2e9859fbbf45ceb09eb6c7e40257fb : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":3},{"id":"4dc80422cfb485a9fb66430f263c285f5f795e6e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-10 09:51:42.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0701958edbc105bc3f597c28d6900820d3a2183b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-10 12:50:06.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e9f1281d361df01e0d793c85d6e3e5cbb8db4e54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-11 15:57:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4549435062e347c48931c4f21bebe7e7eca2f505","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-11 15:57:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a59ac988461377ea9053577d78c45f00fa40ecf0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-11 16:07:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6c701d457012ef889d2bdf266a5a63eb6254945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 16:24:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ad25f0d8a7464b8bdc71661e13cc92\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b2444906cf407ca2f8222f5cfccd9e : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0929857546cf46009e801cb460237b62 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb4795e68a694d6ea1b2b232755e4cfc : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f88d4495ce1e4c149810b9a7fb6f7dc5 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b716b0f50dc4449b104b51436ee4b44 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be46da9260b740d18749d9726281a33a : SUCCESS in 9m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2eeb92494a9641d7b40043740204f49d : SUCCESS in 5m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/50c9d557639245ebb88002d909288e72 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":4},{"id":"5173865ecf04dd01a5c09d5252bc65656882deb7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-12 14:46:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"69544ceb3ecf8fd9bebe819101ca2d15d6ab391d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-01-15 12:54:40.000000000","message":"Patch Set 4: Code-Review+2 PTL-Approved+1 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f7e38849f07dac4e5140aa294096c9f76950aee6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 12:55:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a624196b54a94d0a0eea097d42360374a58ac060","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 13:21:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2c2b52b7acf49daae5fd64b126edeb6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff905adaa284c7ba228c66d29a562fd : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22aeaeac4ed14cb784835ca27de60564 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ec97912032b4abeb968b6bee3e4ba63 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9e1e119718b46c598f0f94a7ee4e6f1 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6719509622145fb898add9c4e7ba9ea : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9981a4969dda4bf685910d5f98d6ee77 : SUCCESS in 8m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e56b0654ef545b593f3ab361e28b428 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":4},{"id":"07ea9c843278cf80b8caa09fff0f1fce461b3e51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 13:21:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ceb33ae249aa6b4b8cd6e9d23230f3d63868c0a8","revisions":{"a6aaae9c7de86b92db6614d2c8226569ff010d65":{"kind":"REWORK","_number":1,"created":"2024-01-04 15:57:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/904767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/904767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/904767/1"}}},"commit":{"parents":[{"commit":"77e8ba577c0a83275d06b6f64c72cdaf1198d05d","subject":"[test] release and cut stable branch for release-test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77e8ba577c0a83275d06b6f64c72cdaf1198d05d"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 12:56:56.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 15:41:29.000000000","tz":60},"subject":"WIP: Introduce EOM tag validation","message":"WIP: Introduce EOM tag validation\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1]. To accomodate this, validator needs to be\nextended to accept \u003cseries\u003e-eom tags that applies at the tip of the\ngiven stable/\u003cseries\u003e branch and the new unmaintained/\u003cseries\u003e branch\nwill be cut from that tag.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: I606ba5e5f8fd783d6c3f83356ded27a46e8bac56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a6aaae9c7de86b92db6614d2c8226569ff010d65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a6aaae9c7de86b92db6614d2c8226569ff010d65"}]},"branch":"refs/heads/master"},"f7fffd5ce462f6df021365000483d99700a2f630":{"kind":"REWORK","_number":2,"created":"2024-01-04 16:22:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/904767/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/904767/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/904767/2"}}},"commit":{"parents":[{"commit":"77e8ba577c0a83275d06b6f64c72cdaf1198d05d","subject":"[test] release and cut stable branch for release-test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77e8ba577c0a83275d06b6f64c72cdaf1198d05d"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 12:56:56.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 16:20:38.000000000","tz":60},"subject":"WIP: Introduce EOM tag validation","message":"WIP: Introduce EOM tag validation\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1]. To accomodate this, validator needs to be\nextended to accept \u003cseries\u003e-eom tags that applies at the tip of the\ngiven stable/\u003cseries\u003e branch and the new unmaintained/\u003cseries\u003e branch\nwill be cut from that tag.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: I606ba5e5f8fd783d6c3f83356ded27a46e8bac56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7fffd5ce462f6df021365000483d99700a2f630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7fffd5ce462f6df021365000483d99700a2f630"}]},"branch":"refs/heads/master"},"e410b8973d81df4f52dd91f6cf6b8f9bbe4f4582":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-01-10 08:53:29.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/904767/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/904767/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/904767/3"}}},"commit":{"parents":[{"commit":"77e8ba577c0a83275d06b6f64c72cdaf1198d05d","subject":"[test] release and cut stable branch for release-test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77e8ba577c0a83275d06b6f64c72cdaf1198d05d"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 12:56:56.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-10 08:53:28.000000000","tz":0},"subject":"Introduce EOM tag validation","message":"Introduce EOM tag validation\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1]. To accomodate this, validator needs to be\nextended to accept \u003cseries\u003e-eom tags that applies at the tip of the\ngiven stable/\u003cseries\u003e branch and the new unmaintained/\u003cseries\u003e branch\nwill be cut from that tag.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: I606ba5e5f8fd783d6c3f83356ded27a46e8bac56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e410b8973d81df4f52dd91f6cf6b8f9bbe4f4582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e410b8973d81df4f52dd91f6cf6b8f9bbe4f4582"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ceb33ae249aa6b4b8cd6e9d23230f3d63868c0a8":{"kind":"REWORK","_number":4,"created":"2024-01-11 15:57:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/904767/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/904767/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/904767/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/904767/4"}}},"commit":{"parents":[{"commit":"77e8ba577c0a83275d06b6f64c72cdaf1198d05d","subject":"[test] release and cut stable branch for release-test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77e8ba577c0a83275d06b6f64c72cdaf1198d05d"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 12:56:56.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-11 15:56:51.000000000","tz":60},"subject":"Introduce EOM tag validation","message":"Introduce EOM tag validation\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1]. To accomodate this, validator needs to be\nextended to accept \u003cseries\u003e-eom tags that applies at the tip of the\ngiven stable/\u003cseries\u003e branch and the new unmaintained/\u003cseries\u003e branch\nwill be cut from that tag.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: I606ba5e5f8fd783d6c3f83356ded27a46e8bac56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ceb33ae249aa6b4b8cd6e9d23230f3d63868c0a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ceb33ae249aa6b4b8cd6e9d23230f3d63868c0a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
