)]}'
{"id":"openstack%2Freleases~904769","triplet_id":"openstack%2Freleases~master~Ia7064e36a2d92ef02f60b7371ebf3e4772abcc80","project":"openstack/releases","branch":"master","topic":"relmgt-unmaintained-state","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2024-01-17 16:16:38.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-01-17 15:44:08.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-01-12 14:44:35.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"Ia7064e36a2d92ef02f60b7371ebf3e4772abcc80","subject":"Add EOM tag option to transition_series.sh","status":"MERGED","created":"2024-01-04 15:57:02.000000000","updated":"2024-01-17 16:16:38.000000000","submitted":"2024-01-17 16:16:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904769-relmgt-unmaintained-state","meta_rev_id":"3178926f49bd5788cf7d7e451105f356819b6877","_number":904769,"virtual_id_number":904769,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-17 16:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2024-01-16 13:02:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-01-11 15:57:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2024-01-12 14:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-01-17 15:44:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-01-15 13:35:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 16:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-05 14:35:56.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2024-01-10 08:58:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-01-10 09:11:10.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-01-16 13:02:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"71d5eb791418d16455591bf475aaa5a64c5e33e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-04 15:57:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87f9ecce25a8378faa1c434a1dfc1d2ac12ec9e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 16:11:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c4c704fb1fb4af1a63e1ba9097d5966\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c01ffa0da9fe4de18bc111d67360df59 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/705e3f5842d049dcbb7ee5df4697b646 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57e7250639b74a41b40d11ed0315c085 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77ea45e50b064a1d8aae5f305c839000 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52bc1810fb384e2f9adce6842ad67857 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4226249fb4496f94e952b766d8912c : SUCCESS in 6m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ce23bba905c49b89e6b14768b144505 : SUCCESS in 5m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a345d802daf647eb87fd3b4150604410 : SUCCESS in 3m 31s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/066e16a22d67433a91a4373a673e169f : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b3b477c3e7c401d9f2b786868c3b97e8336a1371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-04 16:22:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"57873459efe44245631fc1217f3ac29bcf3108e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 16:32:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d598dd01d4c4a1db29d9fa101556024\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76832ec7ffb446a4aae28e08ca2b8798 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57d9a17b752a419282b85478f141ecc1 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94191ccdd1c84ceaae2888e15fa917d2 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6559221faad401e8a1179b2f0fc03b5 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d42b0bb52b984154957be40d0c37ee35 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91437cd099be4ceaa21c1fc168fd955b : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f006b2e00a294621bc8b8631de994172 : SUCCESS in 4m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e1c7231214ec4f868584c77b7e21762b : SUCCESS in 3m 19s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5adcfba88e8c4d10aece75f120e1b20d : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"86241ed9d9868e4f0a513dde1de6ff4d1a52889d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-01-05 14:35:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1664298315fbaf9111bc1ed94c82b263b04a0987","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-10 08:55:12.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3a4e0874f3bdde8fa294fbf1bae4910f504c1dea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-10 08:55:23.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ae0060cf14ae266dbadb39ea2b8153e9e0059cbf","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-01-10 09:11:10.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0f2187f6eab317847bd1295e65e715ef3e4fa57a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 09:31:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8872dd0f1764c128061eab784fd5c2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fadbb906664c4e23aff2a79b1357d168 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a60c349a14fa4a579f7d5281f23bdc1d : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b939d141397244349f65de9a627094ae : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49a7893d547d455bafa78cdec6eb8b42 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1540c250f641405099b40065446ab277 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9a1308a202146209a195b0a469a61cf : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1fe10cf91d64f3ea0b04491f35d3864 : SUCCESS in 4m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8ba0a8c3f65f4ad9939f00794f997f8d : SUCCESS in 3m 23s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/741e7c4ce61d4b8c82aae1a3ad05fd33 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":4},{"id":"524c09201a5c82ac9a874793ccbd07abf8fe05ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-11 15:57:16.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3e9ff3b42b94fa11ce699829ebc734c0a130e86c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 16:10:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87cf619e03bb4b27b10b7edbd3ada4da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244f1fa417f04e27973e30919ec6c964 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29147a4df9974a71b941d91299e5c399 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/597e3e5b56794ed4b8f49fd6cba4cf4c : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f14176c3c2ff4260a59dfc190b1458e5 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0bf399855dde4d7f8a22132990aedbaa : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f2752cad164a24ae945035964400ce : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0814945fd4c4a3fbd39d956adc6ce38 : SUCCESS in 4m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b88d185ffaa440eca6232341c8089933 : SUCCESS in 3m 11s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2b11c30cb5074608bacc1f28f8f3a81f : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":5},{"id":"0cabeee17e1543f1c09aeda78edd2ba376fc2827","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-12 14:44:35.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7a4365026d97cd131ac0e0bf595febc4296d3944","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-15 13:35:08.000000000","message":"Patch Set 5: PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"abbc231b1255f6e8c6ad434d30bb45d0b8c75016","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-16 13:02:30.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6a4307cd4241f842faf4286f1bf155702106a5dd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-17 15:44:08.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0bb47d0f51a0edd8636c7da992c208f7b7dc5be8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 15:44:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2c8832db4b3de9c4ac63f21ad8a04bdce71b2c59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 16:16:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f342ba34136455e9764fc1d7486238d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f8be392ab8242d2903db14912dfdc4e : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16d754c8ec5b4e9593efa50f9e3e04bf : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62e34cfc6e1441259890582ac93e01d8 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3df5ee8b4a2f484dadd1762c90ab2d9e : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdade03c5d5b4cc7aebfc4a23cb73c19 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a1d97fc6674dd3b8595d44cde62032 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/553faef1dcd247ba918aea5a8889eb8b : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":5},{"id":"3178926f49bd5788cf7d7e451105f356819b6877","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 16:16:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4a5c70564b444fcdd558808271056ce14fd76be3","revisions":{"1735636a85d4f377f2923b6995dbc8cf88095847":{"kind":"REWORK","_number":1,"created":"2024-01-04 15:57:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/69/904769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/904769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/904769/1"}}},"commit":{"parents":[{"commit":"16ce098ad0973ed56b7f482ad88230423acb3486","subject":"WIP: Add EOM tag option to new-release command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16ce098ad0973ed56b7f482ad88230423acb3486"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 13:11:11.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 15:41:29.000000000","tz":60},"subject":"WIP: Add EOM tag option to transition_series.sh","message":"WIP: Add EOM tag option to transition_series.sh\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\nThis patch introduces the new \u003cseries\u003e-eom tag option in the\ntransition_series.sh script. The tag applies at the tip of the given\nstable/\u003cseries\u003e branch and the new unmaintained/\u003cseries\u003e branch can be\ncut from that tag.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: Ia7064e36a2d92ef02f60b7371ebf3e4772abcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1735636a85d4f377f2923b6995dbc8cf88095847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1735636a85d4f377f2923b6995dbc8cf88095847"}]},"branch":"refs/heads/master"},"eb464eb5866915a8d9dbd36473861d1044d4f4ff":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-04 16:22:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/69/904769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/904769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/904769/2"}}},"commit":{"parents":[{"commit":"34fd2a1047bc551dce4918a6cb439977aa6852a3","subject":"WIP: Add EOM tag option to new-release command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/34fd2a1047bc551dce4918a6cb439977aa6852a3"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 13:11:11.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 16:21:41.000000000","tz":60},"subject":"WIP: Add EOM tag option to transition_series.sh","message":"WIP: Add EOM tag option to transition_series.sh\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\nThis patch introduces the new \u003cseries\u003e-eom tag option in the\ntransition_series.sh script. The tag applies at the tip of the given\nstable/\u003cseries\u003e branch and the new unmaintained/\u003cseries\u003e branch can be\ncut from that tag.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: Ia7064e36a2d92ef02f60b7371ebf3e4772abcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb464eb5866915a8d9dbd36473861d1044d4f4ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb464eb5866915a8d9dbd36473861d1044d4f4ff"}]},"branch":"refs/heads/master"},"4ceac69613747d421b757d716be4f3aa08671c95":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-01-10 08:55:12.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/69/904769/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/904769/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/904769/3"}}},"commit":{"parents":[{"commit":"34fd2a1047bc551dce4918a6cb439977aa6852a3","subject":"WIP: Add EOM tag option to new-release command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/34fd2a1047bc551dce4918a6cb439977aa6852a3"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 13:11:11.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-10 08:55:12.000000000","tz":0},"subject":"Add EOM tag option to transition_series.sh","message":"Add EOM tag option to transition_series.sh\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\nThis patch introduces the new \u003cseries\u003e-eom tag option in the\ntransition_series.sh script. The tag applies at the tip of the given\nstable/\u003cseries\u003e branch and the new unmaintained/\u003cseries\u003e branch can be\ncut from that tag.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: Ia7064e36a2d92ef02f60b7371ebf3e4772abcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ceac69613747d421b757d716be4f3aa08671c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ceac69613747d421b757d716be4f3aa08671c95"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4fcdc18ef499ef05b06b03af745cc243270a5f02":{"kind":"NO_CHANGE","_number":4,"created":"2024-01-10 08:55:23.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/69/904769/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/904769/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/904769/4"}}},"commit":{"parents":[{"commit":"6470bdd21a530e488e8f7c1c88666c619f1d2af8","subject":"Add EOM tag option to new-release command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6470bdd21a530e488e8f7c1c88666c619f1d2af8"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 13:11:11.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-10 08:55:23.000000000","tz":0},"subject":"Add EOM tag option to transition_series.sh","message":"Add EOM tag option to transition_series.sh\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\nThis patch introduces the new \u003cseries\u003e-eom tag option in the\ntransition_series.sh script. The tag applies at the tip of the given\nstable/\u003cseries\u003e branch and the new unmaintained/\u003cseries\u003e branch can be\ncut from that tag.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: Ia7064e36a2d92ef02f60b7371ebf3e4772abcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4fcdc18ef499ef05b06b03af745cc243270a5f02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4fcdc18ef499ef05b06b03af745cc243270a5f02"}]},"branch":"refs/heads/master","description":"Rebase"},"4a5c70564b444fcdd558808271056ce14fd76be3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-01-11 15:57:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/69/904769/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/904769/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/904769/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/904769/5"}}},"commit":{"parents":[{"commit":"8a9905187557d4f0844a354c8b1eaafe7a7cde5f","subject":"Add EOM tag option to new-release command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a9905187557d4f0844a354c8b1eaafe7a7cde5f"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-04 13:11:11.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-11 15:56:57.000000000","tz":60},"subject":"Add EOM tag option to transition_series.sh","message":"Add EOM tag option to transition_series.sh\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\nThis patch introduces the new \u003cseries\u003e-eom tag option in the\ntransition_series.sh script. The tag applies at the tip of the given\nstable/\u003cseries\u003e branch and the new unmaintained/\u003cseries\u003e branch can be\ncut from that tag.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: Ia7064e36a2d92ef02f60b7371ebf3e4772abcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a5c70564b444fcdd558808271056ce14fd76be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a5c70564b444fcdd558808271056ce14fd76be3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
