)]}'
{"id":"openstack%2Freleases~905154","triplet_id":"openstack%2Freleases~master~I5e4be324118c7c47118ba39a5d29c7046755c9ca","project":"openstack/releases","branch":"master","topic":"d-elections","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2024-03-22 14:52:35.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-03-22 14:52:35.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2024-03-22 14:52:35.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-03-22 14:43:04.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"I5e4be324118c7c47118ba39a5d29c7046755c9ca","subject":"Add 2024.2/\"D\" Election timeline to the schedule","status":"MERGED","created":"2024-01-10 00:32:17.000000000","updated":"2024-03-22 14:52:35.000000000","submitted":"2024-03-22 14:52:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"905154-d-elections","meta_rev_id":"c56da45de48ba9cbd513187b183924148a5dda54","_number":905154,"virtual_id_number":905154,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-22 14:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-03-21 15:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-03-22 14:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-03-22 14:45:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-03-22 14:43:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2024-03-21 15:56:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}],"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 00:54:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-15 12:54:14.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2024-01-15 14:19:38.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-02-28 15:28:00.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"CC"},{"updated":"2024-03-21 15:56:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-03-22 14:45:38.000000000","updated_by":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"reviewer":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"state":"REVIEWER"}],"messages":[{"id":"5cd6a2f098ac9824146391dfa3ebf5531c87b82f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-10 00:32:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7043402a0e7b5765ec020fdbffff9be6e41b5dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 00:54:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44867afe5b304eafb972fbb7fdc97217\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977fbba46617442aa3762bc17f3e8879 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"01c88547b5e49d22d36ac075083d70f86dc0d0f4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-01-15 12:54:14.000000000","message":"Patch Set 1: Code-Review+2 PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"d996b51800b925bf4fa43677ce0ee730d2cbcaf7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-15 14:19:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2a9171a1748bed52a1c8ffcf9c0effbbe53c8ecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-30 10:22:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b1ec2d5f686e7f492df61e27a6e5fdba64ccfdc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 10:36:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9211f157edd3413eb7c95b3030c73c42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c07bdf5b9b40808b3e9b01d76e5979 : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":2},{"id":"3d91b371cce081413317560224558188eaff44dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-28 15:25:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"21f0e364499c8a2c45211c7346c87dbc115bd897","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-28 15:28:00.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5cb8719390beba3133d6c7f871a2242b88d2b6f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 15:35:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b88082e6fa14c7e87c0b944d4c18ac8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a11ce7b7d9fa41c5b50032156781bf0a : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":3},{"id":"b82410770fbe675ccd25dbf413f88ad26d5d82b1","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-28 17:20:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"427da4f677c7937f2be19596a49894fcf5112aef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-21 15:56:08.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4923cc3c60cf73296e8674c0f724046624c2edff","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-22 14:43:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd9a22872f04e72ee2c25ec8547aeae91ec1990f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 14:43:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f4a67b0ec41caa0864ea4f6e4d5d75ac18388b42","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2024-03-22 14:45:38.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"49900fc97c8711b02dc31b7c182ee033327faf2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 14:52:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49278d7aa48d4624b99e5d7a2e49c326\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02892326b05b44f4982ae4c716bdfb3f : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebef844d88ff44de9ac584c735b553af : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":3},{"id":"c56da45de48ba9cbd513187b183924148a5dda54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 14:52:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8510588cab20696d843c0640f2c3718f41ce8f99","revisions":{"ac5ff602a2437a25ba4a9b6144e506caedd7a9c9":{"kind":"REWORK","_number":1,"created":"2024-01-10 00:32:17.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/54/905154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/905154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/905154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/905154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/905154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/905154/1"}}},"commit":{"parents":[{"commit":"275744f5dbf6d5c859c9f469b45a369cff3c8414","subject":"Merge \"Set Stein series state as End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/275744f5dbf6d5c859c9f469b45a369cff3c8414"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-10 00:29:53.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-10 00:29:53.000000000","tz":660},"subject":"Add 2024.2/\"D\" Election timeline to the schedule","message":"Add 2024.2/\"D\" Election timeline to the schedule\n\nWhen trying to add this I noticed that there are duplicate\n\u0027project-specific\u0027 in the schedule for R-7, merge into a single list.\n\nDepends-On: https://review.opendev.org/c/openstack/election/+/905152\nChange-Id: I5e4be324118c7c47118ba39a5d29c7046755c9ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac5ff602a2437a25ba4a9b6144e506caedd7a9c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac5ff602a2437a25ba4a9b6144e506caedd7a9c9"}]},"branch":"refs/heads/master"},"579f99ff9fda8c9f5e92aa659d31bb5b3dcfcc26":{"kind":"REWORK","_number":2,"created":"2024-01-30 10:22:06.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/54/905154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/905154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/905154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/905154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/905154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/905154/2"}}},"commit":{"parents":[{"commit":"a85ab8d4826ff50a6697e7800ce47e78141742e8","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a85ab8d4826ff50a6697e7800ce47e78141742e8"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-10 00:29:53.000000000","tz":660},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-30 10:21:54.000000000","tz":60},"subject":"Add 2024.2/\"D\" Election timeline to the schedule","message":"Add 2024.2/\"D\" Election timeline to the schedule\n\nWhen trying to add this I noticed that there are duplicate\n\u0027project-specific\u0027 in the schedule for R-7, merge into a single list.\n\nDepends-On: https://review.opendev.org/c/openstack/election/+/905152\nChange-Id: I5e4be324118c7c47118ba39a5d29c7046755c9ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/579f99ff9fda8c9f5e92aa659d31bb5b3dcfcc26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/579f99ff9fda8c9f5e92aa659d31bb5b3dcfcc26"}]},"branch":"refs/heads/master"},"8510588cab20696d843c0640f2c3718f41ce8f99":{"kind":"REWORK","_number":3,"created":"2024-02-28 15:25:18.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/54/905154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/905154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/905154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/905154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/905154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/905154/3"}}},"commit":{"parents":[{"commit":"b8835f2d96478a0bfb10abc788fea2a78cc4a96e","subject":"Merge \"[Quality Assurance] Transition Yoga to Unmaintained\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8835f2d96478a0bfb10abc788fea2a78cc4a96e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-10 00:29:53.000000000","tz":660},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-02-28 15:24:46.000000000","tz":540},"subject":"Add 2024.2/\"D\" Election timeline to the schedule","message":"Add 2024.2/\"D\" Election timeline to the schedule\n\nWhen trying to add this I noticed that there are duplicate\n\u0027project-specific\u0027 in the schedule for R-7, merge into a single list.\n\nDepends-On: https://review.opendev.org/c/openstack/election/+/905152\nChange-Id: I5e4be324118c7c47118ba39a5d29c7046755c9ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8510588cab20696d843c0640f2c3718f41ce8f99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8510588cab20696d843c0640f2c3718f41ce8f99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
