)]}'
{"id":"openstack%2Freleases~906562","triplet_id":"openstack%2Freleases~master~Ic444671d8bc472fe745e04804ea399908dd7cfcc","project":"openstack/releases","branch":"master","topic":"yoga-unmaintained","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-02-19 09:02:23.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"6732":{"account":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"last_update":"2024-02-19 09:16:03.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-02-19 09:16:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic444671d8bc472fe745e04804ea399908dd7cfcc","subject":"[trove] Transition Yoga to Unmaintained","status":"MERGED","created":"2024-01-24 22:30:43.000000000","updated":"2024-02-19 09:16:03.000000000","submitted":"2024-02-19 09:16:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906562-yoga-unmaintained","meta_rev_id":"db4d33cca381f8afab18f06b618588818079cc37","_number":906562,"virtual_id_number":906562,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-19 09:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-19 06:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-02-19 09:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-19 09:02:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-02-19 06:29:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-24 22:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-24 22:41:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-24 23:21:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2024-01-24 23:21:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2024-02-08 09:24:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-02-19 06:29:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d7491777ef27e4dcacb8dda5b4f64c0e7a7ebc7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 22:30:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2a0c5940a1c64fee2d8f54f6ca918e0b4dc6708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 22:41:51.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1265461b6222421e90a53b97167fca39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e36ab5176c443c0b8edc9a590f3a4e4 : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/219463f80a014a5e949e78502d2e9eeb : FAILURE in 4m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/413d7ffaf4a44df2a2f81d156390a997 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"10ee597c3ad403ba4beded5855e633ae8ec81b2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 22:50:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8ce8ae99c8e3269aac23d30bd91dceff8a9f54d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 22:59:27.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d368bb4068f4b4abe4d599ae608ebb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8bcbd8806e34193b763b399c98a689a : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1702ce10adba4492b14befa257a8815a : FAILURE in 3m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/364fc0a4f12d4b068e7a6da5c417b507 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"c828635029bd3c894b72d1ce9dd58c93d04feb05","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-25 07:04:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1528d6e09148bb730989b4d6f92e741557ed5aec","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 07:05:36.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9782371947e093d0cd028ec9011b32f31d2954ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-25 12:07:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4be11f9061cdc600c83bfb224c3fa0093d76694c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:19:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d7cfea9e22a4eef8f3c41708f193f72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ec94042ff041b8ac68ea676e561a9c : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62e906eab22f47738f081a4d8642acf8 : SUCCESS in 3m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2f73a80e7f244f16ac6dc7c40fba7932 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"eceb0d086d03ab4ea36c8472cfbb03350dcbfd6b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-08 09:24:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7acceb4cd4e615e838ad30162f25680a4319d1fa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 06:29:20.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b506e07856170a9e0ffa3329d2c191c1d92d1c1f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-19 09:02:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d6fb3c025ea5fedfe0ab7cc82e8a8a8a12d86c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 09:02:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"20594dc2d37516d1a9b3e31b3a1c3d5d87abb32d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 09:16:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/984d92480dae4c0ab21d56c488b1557f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8338338f58e2429787dce89cc2dd6b2e : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e17a3d256d4e4d70b242d70c175c20f8 : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04291542bd2048b7b628084a229815a7 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":3},{"id":"db4d33cca381f8afab18f06b618588818079cc37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 09:16:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4d55bce6cade0336adff599abffdc8ff191ca9dd","revisions":{"b2828f690bc52143572a3b595a7325cba452d66c":{"kind":"REWORK","_number":1,"created":"2024-01-24 22:30:43.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/62/906562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/906562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/906562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/906562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/906562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/906562/1"}}},"commit":{"parents":[{"commit":"017b41c3eae1c7053e0cd585ec8cbae544d26be5","subject":"Merge \"Remove monasca deliverables from 2024.1 Caracal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/017b41c3eae1c7053e0cd585ec8cbae544d26be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:38.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:38.000000000","tz":60},"subject":"[trove] Transition Yoga to Unmaintained","message":"[trove] Transition Yoga to Unmaintained\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\n\nSo this patch transitions stable/yoga branch to Unmaintained as follows:\n- tip of the stable/yoga branch is tagged with yoga-eom tag\n- unmaintained/yoga branch will be cut from yoga-eom tag\n- stable/yoga branch will be deleted\n\nSome further notes:\n- there will no longer be official releases off of yoga branch\n- backports can be pushed against unmaintained/yoga branch\n- openstack-unmaintained-core and \u003cproject\u003e-unmaintained-core groups\n  will have the rights to merge patches\n\nFor further details please read the resolution [1] \u0026 review and +1 this\npatch if this is correct as soon as possible.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: Ic444671d8bc472fe745e04804ea399908dd7cfcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b2828f690bc52143572a3b595a7325cba452d66c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b2828f690bc52143572a3b595a7325cba452d66c"}]},"branch":"refs/heads/master"},"f761948b66d1865973b8aaa26656847b53b75d8a":{"kind":"REWORK","_number":2,"created":"2024-01-24 22:50:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/62/906562/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/906562/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/906562/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/906562/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/906562/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/906562/2"}}},"commit":{"parents":[{"commit":"017b41c3eae1c7053e0cd585ec8cbae544d26be5","subject":"Merge \"Remove monasca deliverables from 2024.1 Caracal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/017b41c3eae1c7053e0cd585ec8cbae544d26be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:38.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:50:53.000000000","tz":60},"subject":"[trove] Transition Yoga to Unmaintained","message":"[trove] Transition Yoga to Unmaintained\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\n\nSo this patch transitions stable/yoga branch to Unmaintained as follows:\n- tip of the stable/yoga branch is tagged with yoga-eom tag\n- unmaintained/yoga branch will be cut from yoga-eom tag\n- stable/yoga branch will be deleted\n\nSome further notes:\n- there will no longer be official releases off of yoga branch\n- backports can be pushed against unmaintained/yoga branch\n- openstack-unmaintained-core and \u003cproject\u003e-unmaintained-core groups\n  will have the rights to merge patches\n\nFor further details please read the resolution [1] \u0026 review and +1 this\npatch if this is correct as soon as possible.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: Ic444671d8bc472fe745e04804ea399908dd7cfcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f761948b66d1865973b8aaa26656847b53b75d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f761948b66d1865973b8aaa26656847b53b75d8a"}]},"branch":"refs/heads/master"},"4d55bce6cade0336adff599abffdc8ff191ca9dd":{"kind":"REWORK","_number":3,"created":"2024-01-25 12:07:10.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/62/906562/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/906562/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/906562/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/906562/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/906562/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/906562/3"}}},"commit":{"parents":[{"commit":"017b41c3eae1c7053e0cd585ec8cbae544d26be5","subject":"Merge \"Remove monasca deliverables from 2024.1 Caracal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/017b41c3eae1c7053e0cd585ec8cbae544d26be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:38.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-25 12:07:08.000000000","tz":60},"subject":"[trove] Transition Yoga to Unmaintained","message":"[trove] Transition Yoga to Unmaintained\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\n\nSo this patch transitions stable/yoga branch to Unmaintained as follows:\n- tip of the stable/yoga branch is tagged with yoga-eom tag\n- unmaintained/yoga branch will be cut from yoga-eom tag\n- stable/yoga branch will be deleted\n\nSome further notes:\n- there will no longer be official releases off of yoga branch\n- backports can be pushed against unmaintained/yoga branch\n- openstack-unmaintained-core and \u003cproject\u003e-unmaintained-core groups\n  will have the rights to merge patches\n\nFor further details please read the resolution [1] \u0026 review and +1 this\npatch if this is correct as soon as possible.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: Ic444671d8bc472fe745e04804ea399908dd7cfcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d55bce6cade0336adff599abffdc8ff191ca9dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d55bce6cade0336adff599abffdc8ff191ca9dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
