)]}'
{"id":"openstack%2Freleases~906569","triplet_id":"openstack%2Freleases~master~I2224695c235c2030c45bbe8ca05e301b9665e63d","project":"openstack/releases","branch":"master","topic":"yoga-unmaintained","attention_set":{},"removed_from_attention_set":{"8064":{"account":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"last_update":"2024-02-19 12:44:48.000000000","reason":"\u003cGERRIT_ACCOUNT_8064\u003e replied on the change","reason_account":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"}},"20498":{"account":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"last_update":"2024-02-20 10:58:39.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-20 10:46:17.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-02-20 10:58:39.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-02-19 15:58:31.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14394":{"account":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"last_update":"2024-02-20 10:58:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2224695c235c2030c45bbe8ca05e301b9665e63d","subject":"[magnum] Transition Yoga to Unmaintained","status":"MERGED","created":"2024-01-24 22:31:20.000000000","updated":"2024-02-20 10:58:39.000000000","submitted":"2024-02-20 10:58:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906569-yoga-unmaintained","meta_rev_id":"9f319f1887cbb7d39d8e551909a7fd64eed7340e","_number":906569,"virtual_id_number":906569,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-20 10:58:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"all":[{"value":1,"date":"2024-02-19 12:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":2,"date":"2024-02-20 10:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-19 15:58:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-02-19 16:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":1,"date":"2024-02-20 10:46:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-02-19 15:59:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"}],"reviewers":{"REVIEWER":[{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-24 22:32:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-24 22:45:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-24 23:20:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"state":"REVIEWER"},{"updated":"2024-01-24 23:20:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-02-08 07:51:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-19 11:58:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2024-02-19 11:58:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"state":"REVIEWER"}],"messages":[{"id":"932685021b70e75709c4b494b303f9dae3e2b369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 22:31:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cfc86eff585fcc218a60d00d552aba34d99af04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 22:45:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7a7c2288e2b40cb9cd853b8f06921a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b93bc87ab841c2a5f1836f1456b286 : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09f492866c2f4dc2a8356e8d73632914 : FAILURE in 5m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/85ed0c1d78a2428da84a3e50858141de : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2cd6f4ae364e81c58be8e27b72714284fbc4e54e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 22:55:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"09d2bf480d40c3700249508de561f06b66ca9d21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 23:08:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b97c3eb705046568b3fa0b87246b4a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a80c87ee8994ecba04b4d28fe68094e : SUCCESS in 6m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4cbd75c3f63f4371859de2175c971609 : SUCCESS in 4m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7ce4277232af432da37a9c33e5b635ba : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"fd8a2d5ead63f82c217db32b1aabd29e4569d216","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-08 07:51:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4d341234c37bfe8a842ff9a2c541126e04bfb93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 06:09:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d43d25a9b2d0b2ed3e6a9d351049ed0de2ab0b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 06:16:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/799f5f03bc5c4feeb4ae0fc84541a105\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9464ccc1eee94504b24e81bf4ccceb3a : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/905b20ed85394fceb6ff57b37629d0da : SUCCESS in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/41a8aeb32aa840cca90df7464f2d38f9 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":3},{"id":"0309fc73c4279cb3ef3a247256f8a506f14c5cf2","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2024-02-19 12:44:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c3e4c34203b7c0b6f2a82837a7aff8bc365bfbaf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-19 15:58:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"76e77307e5d616fb79d46468bc26dc9bb837ad82","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 15:59:21.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"dec464e3cb4ec59d4393fd657d4b67057667c662","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-19 16:05:52.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0483f5552c2e6ce9b5b481c91ba08015485f3d5e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-20 10:46:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5c12040adfea931d344e7f74397e99bc02c5a68d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 10:46:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a0466304ee870388f591551b16f1746bffbf2b75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 10:58:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c1bcb67471e4faa80e5829e4dc8ea8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713e86501ed145579aeca27ef754e568 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487fe4e53552494c98309d3c92866b82 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3a3b38a9a284be7acbc6e92e4f437d1 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":3},{"id":"9f319f1887cbb7d39d8e551909a7fd64eed7340e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 10:58:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aa66ac9ad0a65613c8338ce19865ec83f6246354","revisions":{"45c9011f5fe255dcb51df599c5bca09b65cb1f95":{"kind":"REWORK","_number":1,"created":"2024-01-24 22:31:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/69/906569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/906569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/906569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/906569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/906569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/906569/1"}}},"commit":{"parents":[{"commit":"017b41c3eae1c7053e0cd585ec8cbae544d26be5","subject":"Merge \"Remove monasca deliverables from 2024.1 Caracal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/017b41c3eae1c7053e0cd585ec8cbae544d26be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:39.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:39.000000000","tz":60},"subject":"[magnum] Transition Yoga to Unmaintained","message":"[magnum] Transition Yoga to Unmaintained\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\n\nSo this patch transitions stable/yoga branch to Unmaintained as follows:\n- tip of the stable/yoga branch is tagged with yoga-eom tag\n- unmaintained/yoga branch will be cut from yoga-eom tag\n- stable/yoga branch will be deleted\n\nSome further notes:\n- there will no longer be official releases off of yoga branch\n- backports can be pushed against unmaintained/yoga branch\n- openstack-unmaintained-core and \u003cproject\u003e-unmaintained-core groups\n  will have the rights to merge patches\n\nFor further details please read the resolution [1] \u0026 review and +1 this\npatch if this is correct as soon as possible.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: I2224695c235c2030c45bbe8ca05e301b9665e63d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/45c9011f5fe255dcb51df599c5bca09b65cb1f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/45c9011f5fe255dcb51df599c5bca09b65cb1f95"}]},"branch":"refs/heads/master"},"5b0e394a395514c3202512fbedaf6edd4f757650":{"kind":"REWORK","_number":2,"created":"2024-01-24 22:55:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/69/906569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/906569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/906569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/906569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/906569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/906569/2"}}},"commit":{"parents":[{"commit":"017b41c3eae1c7053e0cd585ec8cbae544d26be5","subject":"Merge \"Remove monasca deliverables from 2024.1 Caracal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/017b41c3eae1c7053e0cd585ec8cbae544d26be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:39.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:55:50.000000000","tz":60},"subject":"[magnum] Transition Yoga to Unmaintained","message":"[magnum] Transition Yoga to Unmaintained\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\n\nSo this patch transitions stable/yoga branch to Unmaintained as follows:\n- tip of the stable/yoga branch is tagged with yoga-eom tag\n- unmaintained/yoga branch will be cut from yoga-eom tag\n- stable/yoga branch will be deleted\n\nSome further notes:\n- there will no longer be official releases off of yoga branch\n- backports can be pushed against unmaintained/yoga branch\n- openstack-unmaintained-core and \u003cproject\u003e-unmaintained-core groups\n  will have the rights to merge patches\n\nFor further details please read the resolution [1] \u0026 review and +1 this\npatch if this is correct as soon as possible.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: I2224695c235c2030c45bbe8ca05e301b9665e63d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5b0e394a395514c3202512fbedaf6edd4f757650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5b0e394a395514c3202512fbedaf6edd4f757650"}]},"branch":"refs/heads/master"},"aa66ac9ad0a65613c8338ce19865ec83f6246354":{"kind":"REWORK","_number":3,"created":"2024-02-19 06:09:19.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/69/906569/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/906569/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/906569/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/906569/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/906569/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/906569/3"}}},"commit":{"parents":[{"commit":"f1920a2055b8f42e9e40d729e659d659dfc0102c","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1920a2055b8f42e9e40d729e659d659dfc0102c"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:39.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-19 06:07:22.000000000","tz":60},"subject":"[magnum] Transition Yoga to Unmaintained","message":"[magnum] Transition Yoga to Unmaintained\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\n\nSo this patch transitions stable/yoga branch to Unmaintained as follows:\n- tip of the stable/yoga branch is tagged with yoga-eom tag\n- unmaintained/yoga branch will be cut from yoga-eom tag\n- stable/yoga branch will be deleted\n\nSome further notes:\n- there will no longer be official releases off of yoga branch\n- backports can be pushed against unmaintained/yoga branch\n- openstack-unmaintained-core and \u003cproject\u003e-unmaintained-core groups\n  will have the rights to merge patches\n\nFor further details please read the resolution [1] \u0026 review and +1 this\npatch if this is correct as soon as possible.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: I2224695c235c2030c45bbe8ca05e301b9665e63d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa66ac9ad0a65613c8338ce19865ec83f6246354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa66ac9ad0a65613c8338ce19865ec83f6246354"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
