)]}'
{"id":"openstack%2Freleases~906581","triplet_id":"openstack%2Freleases~master~I7a793845ad18d31d38c5a5ecc72e29b289738484","project":"openstack/releases","branch":"master","topic":"yoga-unmaintained","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-02-08 14:32:33.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-02-06 17:36:02.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I7a793845ad18d31d38c5a5ecc72e29b289738484","subject":"[swift] Transition Yoga to Unmaintained","status":"MERGED","created":"2024-01-24 22:32:39.000000000","updated":"2024-02-08 14:32:33.000000000","submitted":"2024-02-08 14:32:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906581-yoga-unmaintained","meta_rev_id":"87eca4e2b5f87ab0a59c88677a9282d39e718e53","_number":906581,"virtual_id_number":906581,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-08 14:32:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-07 15:25:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-02-08 14:20:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-06 17:36:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-08 14:20:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-02-06 17:43:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-24 22:33:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-24 22:45:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-24 23:22:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-02-07 15:25:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-08 14:20:35.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"897e398718a6339b5ab8f98dff3308e9909bb31f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 22:32:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdd9e6afad4e4ca28791a20681bae77c976bc742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 22:45:35.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9ffcbaa04244ad7bc227676f1c42dc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977843e6bc23481282bc341a4e0e5b14 : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb855543e4924551abae70d43b39ef10 : FAILURE in 3m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7126fcfbd30d4a69a2db994ef29e8810 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"967d48614d6914f2f305dc5cb3d5d4c1e9db5a8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-25 12:11:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"927808287e58757f98028fe61334e574d872cd09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:25:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d81c86e0a084e8c9e3c750abb0284c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dae5657d6b8423fb4e2e32fd0be018d : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e4e15b9094d494bbaefdcb22d4aaeeb : SUCCESS in 5m 01s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c87f8659a91840ceb15c44e99acdaeed : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ec971cbf7d6f04ef902ef10db298b1a905ef589c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-06 17:36:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fe91e85d6f436a4153e4057d4d10732fb32f2d47","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:43:09.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"a3d8cbd3b17c928cb2c7029ee7c86aeeafd1c2da","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-07 15:25:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"115ce569a3f5e853ab171c1cf421ab49aeb97c23","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-02-08 14:20:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c7fa180204f829f8c5dcd2af0c47c4cc1472381e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:21:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f9bd44cc1b7ce29c8a3774d30ea78f38e89b4c60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:32:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/089af3552b834e70ab93876b9bcd761d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d5fe071ac36439887857f4ee7953dbb : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f148903977c41d5bb02b21177a0cdd0 : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4631d37c5c74592a3a79e177257c238 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"87eca4e2b5f87ab0a59c88677a9282d39e718e53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:32:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"077e17cae828d409f3412f7c9de114c2abe3f693","revisions":{"80220502975dd12baae02b9d4a8c7d7aea0b1139":{"kind":"REWORK","_number":1,"created":"2024-01-24 22:32:39.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/81/906581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/906581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/906581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/906581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/906581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/906581/1"}}},"commit":{"parents":[{"commit":"017b41c3eae1c7053e0cd585ec8cbae544d26be5","subject":"Merge \"Remove monasca deliverables from 2024.1 Caracal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/017b41c3eae1c7053e0cd585ec8cbae544d26be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:41.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:41.000000000","tz":60},"subject":"[swift] Transition Yoga to Unmaintained","message":"[swift] Transition Yoga to Unmaintained\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\n\nSo this patch transitions stable/yoga branch to Unmaintained as follows:\n- tip of the stable/yoga branch is tagged with yoga-eom tag\n- unmaintained/yoga branch will be cut from yoga-eom tag\n- stable/yoga branch will be deleted\n\nSome further notes:\n- there will no longer be official releases off of yoga branch\n- backports can be pushed against unmaintained/yoga branch\n- openstack-unmaintained-core and \u003cproject\u003e-unmaintained-core groups\n  will have the rights to merge patches\n\nFor further details please read the resolution [1] \u0026 review and +1 this\npatch if this is correct as soon as possible.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: I7a793845ad18d31d38c5a5ecc72e29b289738484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/80220502975dd12baae02b9d4a8c7d7aea0b1139"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/80220502975dd12baae02b9d4a8c7d7aea0b1139"}]},"branch":"refs/heads/master"},"077e17cae828d409f3412f7c9de114c2abe3f693":{"kind":"REWORK","_number":2,"created":"2024-01-25 12:11:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/81/906581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/906581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/906581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/906581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/906581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/906581/2"}}},"commit":{"parents":[{"commit":"017b41c3eae1c7053e0cd585ec8cbae544d26be5","subject":"Merge \"Remove monasca deliverables from 2024.1 Caracal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/017b41c3eae1c7053e0cd585ec8cbae544d26be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-24 22:27:41.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-25 12:11:33.000000000","tz":60},"subject":"[swift] Transition Yoga to Unmaintained","message":"[swift] Transition Yoga to Unmaintained\n\nWith the new TC resolution the community replaces Extended Maintenance\nwith Unmaintained status [1].\n\nSo this patch transitions stable/yoga branch to Unmaintained as follows:\n- tip of the stable/yoga branch is tagged with yoga-eom tag\n- unmaintained/yoga branch will be cut from yoga-eom tag\n- stable/yoga branch will be deleted\n\nSome further notes:\n- there will no longer be official releases off of yoga branch\n- backports can be pushed against unmaintained/yoga branch\n- openstack-unmaintained-core and \u003cproject\u003e-unmaintained-core groups\n  will have the rights to merge patches\n\nFor further details please read the resolution [1] \u0026 review and +1 this\npatch if this is correct as soon as possible.\n\n[1] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: I7a793845ad18d31d38c5a5ecc72e29b289738484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/077e17cae828d409f3412f7c9de114c2abe3f693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/077e17cae828d409f3412f7c9de114c2abe3f693"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
