)]}'
{"id":"openstack%2Freleases~906657","triplet_id":"openstack%2Freleases~master~Iac8f54b5b18773f8078c006728523f0486171bea","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-02-01 11:25:27.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-01-26 04:36:29.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2024-02-01 11:09:09.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-02-01 11:25:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac8f54b5b18773f8078c006728523f0486171bea","subject":"[python-manilaclient] New 2023.1 stable release","status":"MERGED","created":"2024-01-25 13:44:06.000000000","updated":"2024-02-01 11:25:27.000000000","submitted":"2024-02-01 11:25:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906657","meta_rev_id":"aa5f0e7eb39e424745feb31d83acd0f311cab7da","_number":906657,"virtual_id_number":906657,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 11:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2024-01-26 04:36:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-02-01 07:11:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-02-01 11:09:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-01-29 13:25:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-01-30 08:11:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-01 11:09:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-01-25 19:32:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 13:45:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-25 14:20:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-01-29 13:25:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-01-30 00:55:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-01-30 00:55:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2024-01-30 08:11:53.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-02-01 07:11:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"4c69051873e24d7350b78368b4a5d47318e83227","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-25 13:44:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be3abf8be71fd9b19ce5e73e867d2caaa13a62ae","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-25 13:44:32.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54130a05835be7531496256c12e1eb7a3e833f6a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 13:45:00.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"ccccdea7f3a6837831b91c81a78a753656c787a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 13:55:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac6ac1261d914c16a0e3eb4db34fff42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766f924aee28482cb0bfef3362248210 : SUCCESS in 6m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f956e92824c44c3ebe1a9527a664f37f : SUCCESS in 6m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9a369432d3544dd2a9ca00fdbd48d910 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d4a32d77c0e82e689cb88c6dfb3021e95878f4c3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-25 18:53:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"edcb3c8b92914f7441ef95a5db75e8efb52386eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-25 19:31:48.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1b015cbc62621921b82d2f8bdbc51cdaf477efcf","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-25 19:32:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"341dbba9234be20b90f814f63f213d6f0f69a5a7","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 19:32:38.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"472241dbf4eb753539f34b3007e4e23463ebff78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 19:47:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3798efb4d2a24d1e826f35f3242051b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11ab1f9acad04bc1bbf4f94e41aeb8a5 : SUCCESS in 7m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d935262ba51143649d07dd03633a9670 : SUCCESS in 6m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f1fc8b361b5347ad9a7afdd573a152ed : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c2ffdb1b4291a4e89e99119e76d3577c0dc5193a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-26 04:36:29.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df575ec1f3bf0d9a6ea23b6a3a48887040798175","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-29 13:25:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a5bf9c7095e80e80e4f90e97f9eac8f1c4659107","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-01-30 08:11:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c37ca802334cde92befa8f759e0fd746237c8ad5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-01 07:11:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"edb8f2a25355f2691e3a4b90a750f3f849290735","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-02-01 11:09:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"922b8558f3a79e142ca2e74e38c996687e6963ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 11:09:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aa4d039ed80eedb5ba9cf64951ccbdfcf5267923","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 11:25:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1871f06265b1481f92a36d82bf8e0b46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a522ca0037c447e9960e54bde007667 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e41732a70894f7da43a8e9f99c594b4 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2b242a910d74fefa32b431795dc26fa : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"aa5f0e7eb39e424745feb31d83acd0f311cab7da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 11:25:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83e405fb5f887b398f5638e8677e50387e6c451f","revisions":{"5e27a1b6c82e9bd6f27a6d9d2122019f87bb7f66":{"kind":"REWORK","_number":1,"created":"2024-01-25 13:44:06.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/57/906657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/906657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/906657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/906657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/906657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/906657/1"}}},"commit":{"parents":[{"commit":"aba4c3b9d6ae57b3f83c44907e8896fa2691e068","subject":"[python-manilaclient] New Bobcat stable release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aba4c3b9d6ae57b3f83c44907e8896fa2691e068"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2024-01-25 13:35:14.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2024-01-25 13:35:14.000000000","tz":-180},"subject":"[python-manilaclient] New 2023.1 stable release","message":"[python-manilaclient] New 2023.1 stable release\n\nFixes include:\n$ git log --oneline 4.4.0..3abf89da1b7106565a833b61d09b59b6be39132d --no-merges\n3abf89d (HEAD, origin/stable/2023.1) Enable the option to return the response body\n6261a1b Fix \"test_share_access_show\" osc functional test\n65826d2 (gerrit/stable/2023.1) [CI] switch to using osc for bootstrap\n\nChange-Id: Iac8f54b5b18773f8078c006728523f0486171bea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e27a1b6c82e9bd6f27a6d9d2122019f87bb7f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e27a1b6c82e9bd6f27a6d9d2122019f87bb7f66"}]},"branch":"refs/heads/master"},"83e405fb5f887b398f5638e8677e50387e6c451f":{"kind":"REWORK","_number":2,"created":"2024-01-25 19:31:48.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/57/906657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/906657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/906657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/906657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/906657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/906657/2"}}},"commit":{"parents":[{"commit":"aba4c3b9d6ae57b3f83c44907e8896fa2691e068","subject":"[python-manilaclient] New Bobcat stable release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aba4c3b9d6ae57b3f83c44907e8896fa2691e068"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2024-01-25 13:35:14.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-01-25 19:31:45.000000000","tz":0},"subject":"[python-manilaclient] New 2023.1 stable release","message":"[python-manilaclient] New 2023.1 stable release\n\nFixes include:\n$ git log --oneline 4.4.0..3abf89da1b7106565a833b61d09b59b6be39132d --no-merges\n3abf89d (HEAD, origin/stable/2023.1) Enable the option to return the response body\n6261a1b Fix \"test_share_access_show\" osc functional test\nffaffe4 Fix default share type resolution in OSC\n65826d2 (gerrit/stable/2023.1) [CI] switch to using osc for bootstrap\n\nChange-Id: Iac8f54b5b18773f8078c006728523f0486171bea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83e405fb5f887b398f5638e8677e50387e6c451f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83e405fb5f887b398f5638e8677e50387e6c451f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
