)]}'
{"id":"openstack%2Freleases~906842","triplet_id":"openstack%2Freleases~master~Ic2aa34f9d07c8fe8bbc659843fd932f9edef09f7","project":"openstack/releases","branch":"master","topic":"zed-stable","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2024-03-05 08:54:17.000000000","reason":"\u003cGERRIT_ACCOUNT_29313\u003e replied on the change","reason_account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-03-05 09:27:51.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-03-05 09:37:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic2aa34f9d07c8fe8bbc659843fd932f9edef09f7","subject":"Release horizon for stable/zed","status":"MERGED","created":"2024-01-26 15:21:39.000000000","updated":"2024-03-05 09:37:39.000000000","submitted":"2024-03-05 09:37:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906842-zed-stable","meta_rev_id":"a47428bc0793724dc68ebaa045b5c82d44aa20bf","_number":906842,"virtual_id_number":906842,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-05 09:37:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2024-03-05 08:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2024-03-05 09:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-03-05 08:53:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2024-03-05 09:27:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-03-05 08:54:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-26 15:22:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-26 15:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-26 16:21:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2024-03-04 14:04:31.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"13624f85174212b405f4a5c41cda9606c5f7f254","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-26 15:21:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c97a46480796663454107dc4d09452614bd44f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 15:35:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/556139a2fffa4454891151bf2b1e08b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a423ac8e682944209e39ea25f80be5a6 : SUCCESS in 7m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87e2da5427a446699491c1657dd3678c : SUCCESS in 7m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/285f5967e24b4e698385839c911b5f45 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d1149405eb8d994da3886877047d8ed77eda1619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-03-04 12:35:15.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9aa5f226d5ff7b6cc9fd164a681c0bc8203cb255","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-03-04 12:35:43.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f683e293a3557911b8efd88fd51d993be8474ef2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 12:36:00.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"d73f286275cee0a6fb6f3f8d551e71576d9764a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 12:43:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5637b191e20946948286278b5c088498\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e078835fd0f44598abcd4a2371120ca2 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a92e413e03fd492483abb220724224cf : FAILURE in 6m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7d856759167e4d91b87f2d4b04e58d56 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8ed0371ce07cfdefe5936f31c3d499236604ec56","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-04 12:43:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78c0b5bd587dbf123d49ae46d3c8eb0a5e6fce0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-03-04 12:48:39.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e73a15206456e83d7a6140c9e9c0518f53197538","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 12:49:32.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"356e6a6c1c02545c046c56246cc04fccf68fb1e2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-03-04 12:50:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa84d0f41c675bd0428cd79d87eeb13faef8f29b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-03-04 12:56:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7a3f8c7d728b2eabe33b29cdd6b57db8dae008f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 13:00:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9755c1d3b948408f97e7eeb26e7947dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0d7c068f0604807bad0998223fe0a88 : SUCCESS in 6m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0abdae2135074db2ad3667240ca462d0 : SUCCESS in 6m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9c5ebf25162f4deb85800e1972db15e1 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":3},{"id":"fb622257049011d448b581e1e09104361d7b3499","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-04 13:13:26.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d549243cf0ea7a31c24248c8cffccb29da56074","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-04 14:04:31.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3233ab08454e432d7295e389c9feb1a599fe7e34","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-05 08:52:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3de3e24901f2ce652fb4b7ce3fe01e890afc659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-05 08:53:06.000000000","message":"Patch Set 4: Commit message was updated.\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"97ac5cb013ecbfe45a3f7fa50adb47c366b6885b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 08:53:10.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"10bf6dfe0ea7073a68113240d5dad5d31ea37a4a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-05 08:53:54.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aa64516b23326628d8dba540ad63e7dd1571a79b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 08:53:57.000000000","message":"Patch Set 4: -PTL-Approved","accounts_in_message":[],"_revision_number":4},{"id":"21aacf1cb814bfef4f7a706450e4ae2f8124a1d4","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-03-05 08:54:17.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"44e25844ee71ecc77a211462cd19005ca0b9569c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 08:54:43.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"9bbd099cd69a4103bc38d425dcc38ae988126eef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 09:09:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d10e17524fd24ef88d00f225265fa1a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e9ddc6147a4a968ce282c6a2d2f53f : SUCCESS in 8m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9f50b0663804a9c9b8a40f8adf91f24 : SUCCESS in 5m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/143336c037a8474e921344f335e5aef5 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":4},{"id":"bf881fdac4228b4422c14f28c4251c211d65025c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-05 09:27:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d6c3a420e7551983529abee9df5d44a238cfbf74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 09:28:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0edcf4efc348c8698dd6f46fe1cf92f138a8061a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 09:37:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61707dbec55a4bc1a5e5acfe04804841\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c78fc1f04c614573bba9b2ee1805a266 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b8779e7a1540a4afb1c6145e617dcf : SUCCESS in 7m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/690675d245d143e89ac58a752397b612 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":4},{"id":"a47428bc0793724dc68ebaa045b5c82d44aa20bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 09:37:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c4e6178d954549068a32fefe60916d99689c1588","revisions":{"5a90c6583721c183d3d0fdb60b9da4732872e4fe":{"kind":"REWORK","_number":1,"created":"2024-01-26 15:21:39.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/42/906842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/906842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/906842/1"}}},"commit":{"parents":[{"commit":"a85ab8d4826ff50a6697e7800ce47e78141742e8","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a85ab8d4826ff50a6697e7800ce47e78141742e8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:21:35.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:21:35.000000000","tz":60},"subject":"Release horizon for stable/zed","message":"Release horizon for stable/zed\n\nThis release picks up new commits to horizon since\nthe last release from stable/zed.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 23.0.0..b1a26caae\n253882cbe Fix manage.py to use python3\ne17d3c8ea Fix \"Set admin password\" checkbox visibility\n4f0d4d161 Fix object-store container toggle for turning public access on/off\na29c93d20 Remove usage of openstack/xstatic-font-awesome\n24c655379 Fix qos and trunks tables\n0cb616d26 Disable form auto-complete on login form\na72e5440a Unskipping test_floatingip_associate_disassociate test case\nb2a9046fb Remove Description from Edit Image Metadata\nd4ff4f3c2 fix(swift): allow object creation\nca5609c9b Make integration tests stop on first failure\ne39b3e2bd Fix instance error with attribute cells.\n13824f3af Fix instance table behavior after deleting an instance\nad778d75c Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value\n25c107d71 Fix delete image tests.\nf5ca3b202 Integration test navigation machinery for Angular pages\n4cc580d94 Pin tox\u003c4 for stable branches testing\n6acabc4f3 Fix flavor id auto generation\n5264497e9 Make readonly metadata properties unable to edit\n4cf74f323 Fix preselection for launch Instance dialog\na6f450050 Imported Translations from Zanata\n761e466e7 Fix rescue of volume-based instances\n494f5baf0 Imported Translations from Zanata\n778a52e66 Fix app cred create without project_id for domain admins\n6e8788580 Fix flavor specification at instance overview page\n34717d905 Imported Translations from Zanata\ne79aa87e1 Don\u0027t try and update port security if its not changing\nb67481019 Update TOX_CONSTRAINTS_FILE for stable/zed\ne51c56fa6 Update .gitreview for stable/zed\n\nChange-Id: Ic2aa34f9d07c8fe8bbc659843fd932f9edef09f7\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a90c6583721c183d3d0fdb60b9da4732872e4fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a90c6583721c183d3d0fdb60b9da4732872e4fe"}]},"branch":"refs/heads/master"},"0a2799c23c0d175413cbdc0de71e09d30aee9155":{"kind":"REWORK","_number":2,"created":"2024-03-04 12:35:15.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/42/906842/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/906842/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/906842/2"}}},"commit":{"parents":[{"commit":"a85ab8d4826ff50a6697e7800ce47e78141742e8","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a85ab8d4826ff50a6697e7800ce47e78141742e8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:21:35.000000000","tz":60},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2024-03-04 12:35:15.000000000","tz":0},"subject":"Release horizon for stable/zed","message":"Release horizon for stable/zed\n\nThis release picks up new commits to horizon since\nthe last release from stable/zed.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 23.0.0..b1a26caae\n253882cbe Fix manage.py to use python3\ne17d3c8ea Fix \"Set admin password\" checkbox visibility\n4f0d4d161 Fix object-store container toggle for turning public access on/off\na29c93d20 Remove usage of openstack/xstatic-font-awesome\n24c655379 Fix qos and trunks tables\n0cb616d26 Disable form auto-complete on login form\na72e5440a Unskipping test_floatingip_associate_disassociate test case\nb2a9046fb Remove Description from Edit Image Metadata\nd4ff4f3c2 fix(swift): allow object creation\nca5609c9b Make integration tests stop on first failure\ne39b3e2bd Fix instance error with attribute cells.\n13824f3af Fix instance table behavior after deleting an instance\nad778d75c Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value\n25c107d71 Fix delete image tests.\nf5ca3b202 Integration test navigation machinery for Angular pages\n4cc580d94 Pin tox\u003c4 for stable branches testing\n6acabc4f3 Fix flavor id auto generation\n5264497e9 Make readonly metadata properties unable to edit\n4cf74f323 Fix preselection for launch Instance dialog\na6f450050 Imported Translations from Zanata\n761e466e7 Fix rescue of volume-based instances\n494f5baf0 Imported Translations from Zanata\n778a52e66 Fix app cred create without project_id for domain admins\n6e8788580 Fix flavor specification at instance overview page\n34717d905 Imported Translations from Zanata\ne79aa87e1 Don\u0027t try and update port security if its not changing\nb67481019 Update TOX_CONSTRAINTS_FILE for stable/zed\ne51c56fa6 Update .gitreview for stable/zed\n\nChange-Id: Ic2aa34f9d07c8fe8bbc659843fd932f9edef09f7\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a2799c23c0d175413cbdc0de71e09d30aee9155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a2799c23c0d175413cbdc0de71e09d30aee9155"}]},"branch":"refs/heads/master"},"d062ede355323302f3e5165336147beed2550540":{"kind":"REWORK","_number":3,"created":"2024-03-04 12:48:39.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/42/906842/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/906842/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/906842/3"}}},"commit":{"parents":[{"commit":"a85ab8d4826ff50a6697e7800ce47e78141742e8","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a85ab8d4826ff50a6697e7800ce47e78141742e8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:21:35.000000000","tz":60},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2024-03-04 12:48:39.000000000","tz":0},"subject":"Release horizon for stable/zed","message":"Release horizon for stable/zed\n\nThis release picks up new commits to horizon since\nthe last release from stable/zed.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 23.0.0..b1a26caae\n253882cbe Fix manage.py to use python3\ne17d3c8ea Fix \"Set admin password\" checkbox visibility\n4f0d4d161 Fix object-store container toggle for turning public access on/off\na29c93d20 Remove usage of openstack/xstatic-font-awesome\n24c655379 Fix qos and trunks tables\n0cb616d26 Disable form auto-complete on login form\na72e5440a Unskipping test_floatingip_associate_disassociate test case\nb2a9046fb Remove Description from Edit Image Metadata\nd4ff4f3c2 fix(swift): allow object creation\nca5609c9b Make integration tests stop on first failure\ne39b3e2bd Fix instance error with attribute cells.\n13824f3af Fix instance table behavior after deleting an instance\nad778d75c Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value\n25c107d71 Fix delete image tests.\nf5ca3b202 Integration test navigation machinery for Angular pages\n4cc580d94 Pin tox\u003c4 for stable branches testing\n6acabc4f3 Fix flavor id auto generation\n5264497e9 Make readonly metadata properties unable to edit\n4cf74f323 Fix preselection for launch Instance dialog\na6f450050 Imported Translations from Zanata\n761e466e7 Fix rescue of volume-based instances\n494f5baf0 Imported Translations from Zanata\n778a52e66 Fix app cred create without project_id for domain admins\n6e8788580 Fix flavor specification at instance overview page\n34717d905 Imported Translations from Zanata\ne79aa87e1 Don\u0027t try and update port security if its not changing\nb67481019 Update TOX_CONSTRAINTS_FILE for stable/zed\ne51c56fa6 Update .gitreview for stable/zed\n\nChange-Id: Ic2aa34f9d07c8fe8bbc659843fd932f9edef09f7\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d062ede355323302f3e5165336147beed2550540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d062ede355323302f3e5165336147beed2550540"}]},"branch":"refs/heads/master"},"c4e6178d954549068a32fefe60916d99689c1588":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-03-05 08:53:06.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/42/906842/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/906842/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/906842/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/906842/4"}}},"commit":{"parents":[{"commit":"a85ab8d4826ff50a6697e7800ce47e78141742e8","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a85ab8d4826ff50a6697e7800ce47e78141742e8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:21:35.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-03-05 08:53:06.000000000","tz":0},"subject":"Release horizon for stable/zed","message":"Release horizon for stable/zed\n\nThis release picks up new commits to horizon since\nthe last release from stable/zed.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nChange-Id: Ic2aa34f9d07c8fe8bbc659843fd932f9edef09f7\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c4e6178d954549068a32fefe60916d99689c1588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c4e6178d954549068a32fefe60916d99689c1588"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
