)]}'
{"id":"openstack%2Freleases~906847","triplet_id":"openstack%2Freleases~master~I72463d73609d82011619f123745a9797b9ad6f7d","project":"openstack/releases","branch":"master","topic":"zed-stable","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-02-08 14:32:04.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-02-08 14:32:04.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2024-01-29 13:13:42.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2024-02-08 14:32:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72463d73609d82011619f123745a9797b9ad6f7d","subject":"Release ironic for stable/zed","status":"MERGED","created":"2024-01-26 15:31:36.000000000","updated":"2024-02-08 14:32:04.000000000","submitted":"2024-02-08 14:32:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906847-zed-stable","meta_rev_id":"ee9715a125e50ddcfe6c48e7dffe9d65b2541a1f","_number":906847,"virtual_id_number":906847,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-08 14:32:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2024-02-08 14:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-01-29 14:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-01-29 13:13:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2024-02-08 14:13:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-01-29 13:14:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-26 15:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-26 15:44:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-26 16:21:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-01-26 16:21:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2024-01-26 16:21:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-02-08 14:13:27.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"61330a4868926f4bf1c8e1c118ad679bdac7c5c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-26 15:31:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d8796b42bb38a7325727d387da0667c19834b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 15:44:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e775816d11434511beffc569fb461f83\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae603af7fc64ced8571276e8cde2b1b : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21b534dc9a474128bf08e38f803fc584 : FAILURE in 3m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a8e4910e0a9049b4b3611d9f322c8a64 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c56861c124b50e76885f87d23e2ac63701565d07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-29 12:54:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"40f6f69c65836067ed32220b0ba5bb751c1ea8bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-29 12:58:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"409147fd3f6d0f94ab6e484c344248ff8b10e068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 13:12:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/135d958741de4b0ab276023dd23a07a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ed1f3421f74fd9b470353d0718c348 : SUCCESS in 7m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3cc2bce1d7274141b31902c35b278a47 : SUCCESS in 6m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5f70f87710974d1d824179f0edde18ff : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"10d5b050cfdd9fa434b0d9985fbfe6f1a0490245","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-01-29 13:13:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7d319781446e7e8298f4b96008c13390ee1a943a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 13:14:29.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"b2d74125e6880a0a599d24bf43e2b6bc0d6c1ce6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-29 14:05:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2208c2c93e6377afcb954ee4ddb2debd84fb8bd3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-02-08 14:13:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6272974da1c59bad4f3cf0bae928959e21f90197","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:14:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c0bfc8423ccaad7f87d5b14d3180f2aa647655b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:32:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebfb885f8c094b6e9342c689f65c8944\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e8da6e0cb64b088aca0a4d8f9dcd5a : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bba386a6dfe844ea820851aa32d22834 : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b4db319094a476c9e458c66db2683b5 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":3},{"id":"ee9715a125e50ddcfe6c48e7dffe9d65b2541a1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:32:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5d4b028a9851bc8fac30277db71c5f01c39d9f0b","revisions":{"f002fce6b76e5ec39fb40c522bb7f57d37c52634":{"kind":"REWORK","_number":1,"created":"2024-01-26 15:31:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/47/906847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/906847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/906847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/906847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/906847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/906847/1"}}},"commit":{"parents":[{"commit":"a85ab8d4826ff50a6697e7800ce47e78141742e8","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a85ab8d4826ff50a6697e7800ce47e78141742e8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:31:33.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:31:33.000000000","tz":60},"subject":"Release ironic for stable/zed","message":"Release ironic for stable/zed\n\nThis release picks up new commits to ironic since\nthe last release from stable/zed.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 21.1.0..e38735cb9\ne38735cb9 Use per-node external_http_url for boot ISO\nba97ec80e redfish_address - wrap_ipv6 address\n1bd7ab875 [iRMC] Fix parse_driver_info bug enforcing SNMP v3 under FIPS mode\n4fdf65ca5 Fix self_owned_node policy check\n5e2d72cbf [ci] [stable-only] Cinder fixed; make BFV job vote\n907f71742 [stable-only] [CI] BFV, RBAC jobs marked non-voting\n07497e1b0 Fix Cinder Integration fallout from CVE-2023-2088\nc45cf2017 Handle MissingAttributeError when using OOB inspections to fetch MACs\nec6c37579 Always fall back from hard linking to copying files\nd6b970995 Add error logging on lookup failures in the API\n2b0a5575c Fix online upgrades for Bios/Traits\n4481031d7 Wipe Agent Token when cleaning timeout occcurs\n3d32729fb Clean out agent token even if power is already off\n3e3d4e1cb Do not recalculate checksum if disk_format is not changed\na4a3b31d4 Do not move nodes to CLEAN FAILED with empty last_error\ne928980e7 [iRMC] Handle IPMI incompatibility in iRMC S6 2.x\nb91d174eb Move and fix reno config for releasenotes job\n728301fc2 Fix selinux context of published image hardlink\nbe92735d7 Fix \"\u0027NoneType\u0027 object is not iterable\" in RAID\n1ba3d29a0 Fixes for tox 4.0\n49dade88c Fix unit tests for Python 3.11\nf882fa0fa Align iRMC driver with Ironic\u0027s default boot_mode\nff06b3d2a Fix the invalid glance client test\n17fb4f706 Update TOX_CONSTRAINTS_FILE for stable/zed\ne45428d35 Update .gitreview for stable/zed\n\nChange-Id: I72463d73609d82011619f123745a9797b9ad6f7d\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f002fce6b76e5ec39fb40c522bb7f57d37c52634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f002fce6b76e5ec39fb40c522bb7f57d37c52634"}]},"branch":"refs/heads/master"},"8d68c1faa24d7261e9f0710f09fdc5ed23c8878f":{"kind":"REWORK","_number":2,"created":"2024-01-29 12:54:59.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/47/906847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/906847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/906847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/906847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/906847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/906847/2"}}},"commit":{"parents":[{"commit":"a85ab8d4826ff50a6697e7800ce47e78141742e8","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a85ab8d4826ff50a6697e7800ce47e78141742e8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:31:33.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-01-29 12:54:58.000000000","tz":60},"subject":"Release ironic for stable/zed","message":"Release ironic for stable/zed\n\nThis release picks up new commits to ironic since\nthe last release from stable/zed.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 21.1.0..e38735cb9\ne38735cb9 Use per-node external_http_url for boot ISO\nba97ec80e redfish_address - wrap_ipv6 address\n1bd7ab875 [iRMC] Fix parse_driver_info bug enforcing SNMP v3 under FIPS mode\n4fdf65ca5 Fix self_owned_node policy check\n5e2d72cbf [ci] [stable-only] Cinder fixed; make BFV job vote\n907f71742 [stable-only] [CI] BFV, RBAC jobs marked non-voting\n07497e1b0 Fix Cinder Integration fallout from CVE-2023-2088\nc45cf2017 Handle MissingAttributeError when using OOB inspections to fetch MACs\nec6c37579 Always fall back from hard linking to copying files\nd6b970995 Add error logging on lookup failures in the API\n2b0a5575c Fix online upgrades for Bios/Traits\n4481031d7 Wipe Agent Token when cleaning timeout occcurs\n3d32729fb Clean out agent token even if power is already off\n3e3d4e1cb Do not recalculate checksum if disk_format is not changed\na4a3b31d4 Do not move nodes to CLEAN FAILED with empty last_error\ne928980e7 [iRMC] Handle IPMI incompatibility in iRMC S6 2.x\nb91d174eb Move and fix reno config for releasenotes job\n728301fc2 Fix selinux context of published image hardlink\nbe92735d7 Fix \"\u0027NoneType\u0027 object is not iterable\" in RAID\n1ba3d29a0 Fixes for tox 4.0\n49dade88c Fix unit tests for Python 3.11\nf882fa0fa Align iRMC driver with Ironic\u0027s default boot_mode\nff06b3d2a Fix the invalid glance client test\n17fb4f706 Update TOX_CONSTRAINTS_FILE for stable/zed\ne45428d35 Update .gitreview for stable/zed\n\nChange-Id: I72463d73609d82011619f123745a9797b9ad6f7d\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d68c1faa24d7261e9f0710f09fdc5ed23c8878f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d68c1faa24d7261e9f0710f09fdc5ed23c8878f"}]},"branch":"refs/heads/master"},"5d4b028a9851bc8fac30277db71c5f01c39d9f0b":{"kind":"REWORK","_number":3,"created":"2024-01-29 12:58:05.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/47/906847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/906847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/906847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/906847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/906847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/906847/3"}}},"commit":{"parents":[{"commit":"a85ab8d4826ff50a6697e7800ce47e78141742e8","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a85ab8d4826ff50a6697e7800ce47e78141742e8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:31:33.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-29 12:58:05.000000000","tz":0},"subject":"Release ironic for stable/zed","message":"Release ironic for stable/zed\n\nThis release picks up new commits to ironic since\nthe last release from stable/zed.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 21.1.0..e38735cb9\ne38735cb9 Use per-node external_http_url for boot ISO\nba97ec80e redfish_address - wrap_ipv6 address\n1bd7ab875 [iRMC] Fix parse_driver_info bug enforcing SNMP v3 under FIPS mode\n4fdf65ca5 Fix self_owned_node policy check\n5e2d72cbf [ci] [stable-only] Cinder fixed; make BFV job vote\n907f71742 [stable-only] [CI] BFV, RBAC jobs marked non-voting\n07497e1b0 Fix Cinder Integration fallout from CVE-2023-2088\nc45cf2017 Handle MissingAttributeError when using OOB inspections to fetch MACs\nec6c37579 Always fall back from hard linking to copying files\nd6b970995 Add error logging on lookup failures in the API\n2b0a5575c Fix online upgrades for Bios/Traits\n4481031d7 Wipe Agent Token when cleaning timeout occcurs\n3d32729fb Clean out agent token even if power is already off\n3e3d4e1cb Do not recalculate checksum if disk_format is not changed\na4a3b31d4 Do not move nodes to CLEAN FAILED with empty last_error\ne928980e7 [iRMC] Handle IPMI incompatibility in iRMC S6 2.x\nb91d174eb Move and fix reno config for releasenotes job\n728301fc2 Fix selinux context of published image hardlink\nbe92735d7 Fix \"\u0027NoneType\u0027 object is not iterable\" in RAID\n1ba3d29a0 Fixes for tox 4.0\n49dade88c Fix unit tests for Python 3.11\nf882fa0fa Align iRMC driver with Ironic\u0027s default boot_mode\nff06b3d2a Fix the invalid glance client test\n17fb4f706 Update TOX_CONSTRAINTS_FILE for stable/zed\ne45428d35 Update .gitreview for stable/zed\n\nChange-Id: I72463d73609d82011619f123745a9797b9ad6f7d\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d4b028a9851bc8fac30277db71c5f01c39d9f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d4b028a9851bc8fac30277db71c5f01c39d9f0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
