)]}'
{"id":"openstack%2Freleases~906863","triplet_id":"openstack%2Freleases~master~Ibced3cb3fbec76333b7aed03e41b486006f81035","project":"openstack/releases","branch":"master","topic":"zed-stable","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-04-05 14:30:04.000000000","reason":"Change was abandoned"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-04-05 14:30:04.000000000","reason":"Change was abandoned"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-04-05 14:30:04.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-04-05 14:30:04.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-02-08 09:25:53.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-04-05 14:30:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibced3cb3fbec76333b7aed03e41b486006f81035","subject":"Release neutron for stable/zed","status":"ABANDONED","created":"2024-01-26 15:55:26.000000000","updated":"2024-04-23 15:39:21.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4c424f5a2238a30647b1c4aff767b07957f6ffb0","_number":906863,"virtual_id_number":906863,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-08 07:32:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2024-02-08 07:20:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2024-02-08 09:25:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-26 15:56:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-26 16:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-26 16:20:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-01-26 16:20:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-01-29 13:30:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-01-30 14:25:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-01-30 14:59:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"bc9916ffb8b2071d1dd11b37612363548e193a47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-26 15:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fb409691fcacfcd4ee256b0397e53d3ac8477cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 16:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c7bcc97b19342e3a16ffd3735d37996\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb08e81d301e4634bc4079140e053c34 : SUCCESS in 8m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31b79dddbba948479c0dbbb77876afc7 : SUCCESS in 6m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6aa8ea7d34594e1fb06eacec4c648822 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ceebf9056e48567fdb861f60b45f2612f5ca28dc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-29 13:30:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b550063edf8ad91c918570140673fb85eed8cd1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-07 16:14:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f47756c557c940da4bc4d0937eb76e97443ac923","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-07 16:19:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1977224e95cac456665ed93c9405b84572ddc651","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-08 07:16:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e7989cacebcb53d9b04cd87272caa7d1dec850e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-08 07:19:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2eeb7969b110a9faae6d1151382f76fb90d8f5f2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-08 07:20:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1733b19191d6195a1e445b1f2f9cb9bc277e6325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 07:32:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fffeca00119e461e89e01e11901f26a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a84222beb43742b0b5ed52a874c2d553 : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/083c336008df4013a6ba8b4cbad7eb49 : SUCCESS in 7m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/330b7b79ac1a4465b112b6816ee4c321 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"40be18c00f0b6c4a8d299b662780092c12d0e4b0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-08 09:25:53.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"764a2b57564574a308acd6044a61f6bae617eaf4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-05 14:30:04.000000000","message":"Abandoned\n\nthis seems to got stuck, abandoning. please propose a new patch if a release is needed before the transition.","accounts_in_message":[],"_revision_number":2},{"id":"4fa97af0d10fe58e4657c7a79afff1b21db94567","tag":"autogenerated:gerrit:restore","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-23 14:45:43.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"7d273089d8ff4ee61f41ef5384f23346336665e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:03:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68f1b89cf69146f7ae8e63d1d4f7f436\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a592bd57d9545759c0653d486631841 : SUCCESS in 7m 50s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/9ab661ed8040435c889e26c0c71a5621 : SUCCESS in 8m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/32977acf787740eca45b561d7a30acc2 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4c424f5a2238a30647b1c4aff767b07957f6ffb0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-23 15:39:21.000000000","message":"Abandoned\n\nAbandoning in favor of https://review.opendev.org/c/openstack/releases/+/916793","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"853ae9dd9b9784ea46e5ac054997e6da303b658b","revisions":{"3430b34f01d5cc5bfea046d5eb73a976bb1c38f8":{"kind":"REWORK","_number":1,"created":"2024-01-26 15:55:26.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/63/906863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/63/906863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/906863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/906863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/906863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/63/906863/1"}}},"commit":{"parents":[{"commit":"a85ab8d4826ff50a6697e7800ce47e78141742e8","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a85ab8d4826ff50a6697e7800ce47e78141742e8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:55:22.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:55:22.000000000","tz":60},"subject":"Release neutron for stable/zed","message":"Release neutron for stable/zed\n\nThis release picks up new commits to neutron since\nthe last release from stable/zed.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 21.2.0..0ae1d93932\n0396b33bc4 Revert \"Add sleep before checking if ovs port is in the namespace\"\n905e2e430a Forbid the subnet gateway IP deletion if a router interface is attached\n8f15164068 Register Chassis_Private table in BaseOvnSbIdl\n222c997022 Add sleep before checking if ovs port is in the namespace\n2c8076dd28 Add max limit to agent_down_time\nbbe235779b Improve the SG RPC callback ``security_group_info_for_ports``\n602bb1a325 [OVN][FT] Make explicit the \"publish\" call check in \"test_port_forwarding\"\n4eb234a447 [OVN] DB sync host/physnet - filter on agent_type\n4698dd899f Metadata: handle process exceptions\n74a3262a8a Correctly validate subnet arguments when using a subnetpool\nde5b7244ee [Fullstack] Double check that agent is dead when it should be dead\n69e0dacd05 [DHCP agent] Fetch OVN Metadata port from plugin\n9d22276013 get_hosts_mapped_with_segments add filter agt_type\ncbebc3dd8a [DHCP agent] Fix route to OVN metadata port for non-isolated networks\nb2100c6a29 docs: update default value of metadata workers for ml2/ovn\n61ad633a33 Catch non-existent entry failures better in ip_lib\n\nChange-Id: Ibced3cb3fbec76333b7aed03e41b486006f81035\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3430b34f01d5cc5bfea046d5eb73a976bb1c38f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3430b34f01d5cc5bfea046d5eb73a976bb1c38f8"}]},"branch":"refs/heads/master"},"853ae9dd9b9784ea46e5ac054997e6da303b658b":{"kind":"REWORK","_number":2,"created":"2024-02-08 07:19:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/906863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/63/906863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/906863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/906863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/906863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/63/906863/2"}}},"commit":{"parents":[{"commit":"2d729d808599c0e66b47cdf66ebbcd8d569940f1","subject":"Merge \"[OpenStackSDK] Transition Ussuri to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d729d808599c0e66b47cdf66ebbcd8d569940f1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-26 15:55:22.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-08 07:19:41.000000000","tz":0},"subject":"Release neutron for stable/zed","message":"Release neutron for stable/zed\n\nThis release picks up new commits to neutron since\nthe last release from stable/zed.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 21.2.0..f83e1fd4f0\n1f5ce18a14 [OVN][FT] Retry in case of timeout when executing \"ovsdb-client\".\nc47c600b6a [Fullstack] Consolidate segmentation_id update tests into single test\n5250b62a85 [Fullstack] Remove SecurityGroupRulesTest.test_normalized_cidr_in_rule test\naf83df07f4 [Fullstack] Remove test_port_shut_down module\n69bda7333f [Fullstack] Remove unnecessary tests\n2de488ab11 Set trunk parent port as access port in ovs to avoid loop\n0396b33bc4 Revert \"Add sleep before checking if ovs port is in the namespace\"\n905e2e430a Forbid the subnet gateway IP deletion if a router interface is attached\n8f15164068 Register Chassis_Private table in BaseOvnSbIdl\n222c997022 Add sleep before checking if ovs port is in the namespace\n2c8076dd28 Add max limit to agent_down_time\n5dcad1d7c7 Don\u0027t set port capabilities for OVS HW offloading\nbbe235779b Improve the SG RPC callback ``security_group_info_for_ports``\n602bb1a325 [OVN][FT] Make explicit the \"publish\" call check in \"test_port_forwarding\"\n4eb234a447 [OVN] DB sync host/physnet - filter on agent_type\n4698dd899f Metadata: handle process exceptions\n74a3262a8a Correctly validate subnet arguments when using a subnetpool\nde5b7244ee [Fullstack] Double check that agent is dead when it should be dead\n69e0dacd05 [DHCP agent] Fetch OVN Metadata port from plugin\n9d22276013 get_hosts_mapped_with_segments add filter agt_type\ncbebc3dd8a [DHCP agent] Fix route to OVN metadata port for non-isolated networks\nb2100c6a29 docs: update default value of metadata workers for ml2/ovn\n61ad633a33 Catch non-existent entry failures better in ip_lib\n\nChange-Id: Ibced3cb3fbec76333b7aed03e41b486006f81035\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/853ae9dd9b9784ea46e5ac054997e6da303b658b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/853ae9dd9b9784ea46e5ac054997e6da303b658b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
