)]}'
{"id":"openstack%2Freleases~907502","triplet_id":"openstack%2Freleases~master~Iaaa9702c5b304152492ed2afc56ac07fb9c9fb98","project":"openstack/releases","branch":"master","topic":"relmgt-unmaintained-state","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2024-02-12 13:45:07.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-08 07:17:46.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-02-12 13:19:20.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"Iaaa9702c5b304152492ed2afc56ac07fb9c9fb98","subject":"Add list_eom_stale_branches.sh","status":"MERGED","created":"2024-02-01 22:23:09.000000000","updated":"2024-02-12 13:45:07.000000000","submitted":"2024-02-12 13:45:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907502-relmgt-unmaintained-state","meta_rev_id":"f101050d0bcfa195ec32e3452df75324bafbb038","_number":907502,"virtual_id_number":907502,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-12 13:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-12 13:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-02-12 13:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2024-02-12 13:19:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-12 13:19:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 22:39:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 15:19:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-08 14:22:54.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"beb5c1ee0938a59b13bdd4d2a4aeaf62289914ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-01 22:23:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6ab473f43cf5d437390c7403101777c9db73071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 22:39:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8181142861974f728c9378df092e9fa4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eaf7cd322b44560a0e426165691ac96 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2f6a822cecc4d529d9dc973b5593fb1 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/745dd5f71a9841c191d550d2b9f09560 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a458b0c52421482494f4fa7ef88c2d78 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb2b68ce74174e3d8aaa233d16f99c3c : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48fc230198a8471cbbd04b43b6db4dd5 : SUCCESS in 6m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3aa142f8ac1412e87e9f0d04e072dee : SUCCESS in 4m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/65271d958bef4850bbcaa223683fb00d : SUCCESS in 3m 32s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b2bab13d46304b44a753cd757a8eb6ee : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6f5d96269773c62cba9f7df5a706911d54e06253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-07 15:00:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"08ee1e875865d6766a59741da93ee0997821db63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 15:13:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9de98aa2bcb7473e93e79e5e9a988046\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7ad11584dc47e4b43857c606b668d1 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6a8b621a11c465c8c9706f52cc1c1ba : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a93b51e95c34822a3eb4c1b822c06ea : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1eb9ab8658e43249c04e7997a687708 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c24ebccb357498da3caba4472e7ae6b : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d0f527e1eed400580d0b03da5d87e47 : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66e96482063446fea3e65a2cea7e6f10 : SUCCESS in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6cbe3ae30a964ae78f50b25220ed06a8 : SUCCESS in 3m 35s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a37244677c5c4933a1d704ea6db7a860 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c9fa39357f9ddea899e439b1337fcbc26f9c9de7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-07 15:19:51.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ee9d05f9d7b58e4e6845983938c8f987dfb962e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-07 16:41:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"dc6bfd596c48f3155977b0e118e7fdf516473ca7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-07 16:43:35.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8b9061513a71e6eac792e57b63f6bc4444f7b380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 16:57:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb078f5691114cd1aa18fc41f1ab5d38\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b269d9009044e6ad003c2df7e1e82e : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/198e63ded44249a598bafa8917ad84cb : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c63e70fa0b6d4dee89d28efda46977f0 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc57e9c78f6445dc80ff073dc8500a03 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63785bc80ae84a408a174cd64be016f5 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fadb12dd7a8f4f9b950fabdfb98b7a34 : SUCCESS in 6m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62924944153649bab93b0fc46a218ddb : SUCCESS in 3m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/93e866a76cf1463c87e5c5bf8d5f5e32 : SUCCESS in 3m 26s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/787aef9890494ae0a6c074dfa62a5139 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":3},{"id":"b1f47bba3ca391cad84074de31e98564bc863e9a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-08 07:17:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cf981e3d57c717a7e59ad913768745ba96c69634","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-02-08 14:22:54.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a13aa7c9c2685894c3d1d5e4e30c6b54ed1cfd80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-12 13:19:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"01f390913e4d7c9159072edb152a91b9b83aa5ba","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-12 13:19:20.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eed7ff90ca1a82db4fe0fb718a3f00709e4872f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 13:30:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c82a59366fd4f20b4d2fe367dc2e033\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18f0a584f7174657ad7b5c998e88d144 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0466ecb07fdf4c5c9f176231afdb0e76 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/322d1445c282482abc34fd910e8a1569 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5617d6d381e041d2b4be906258f7b177 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e40a3a1540c44e9a6ba3e00637112b9 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/413c05321328494c80e94bbd662635e5 : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/036f796583ef4111ae006f07e7d2f02c : SUCCESS in 4m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/35dca17a6eeb41e38ebef7bbad54c37e : SUCCESS in 2m 55s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c1f96b6613c54154885b11af79294288 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":4},{"id":"9b38c63bcd6460f5ee3a0f898c2125c88f5c9557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 13:30:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"23c7fde72cb2310346f98116983b62507018c6c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 13:45:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0af48b6725a40c0a2ce044078a2a191\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f56bc104564407873460676f0054ca : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89cb0e72d53944e79ed695c0363d3fd7 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bab2d8b80ae14666afc49f9852322cfd : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f06fc840e8b426c8808a53c6ef0f39a : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20fd2e27b15e47ff9b8e2c91496c9c9e : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba10dc685d34f79b145635985a1ec5f : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64159722d18048ecb6ce85335351e955 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":4},{"id":"f101050d0bcfa195ec32e3452df75324bafbb038","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 13:45:07.000000000","message":"Change has been successfully merged\n\n3 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"97e76562e055747edb3cf729ec9c613fa3f9b678","revisions":{"09ef3cee3767b24eeab5593d74cf3b3a89210a3b":{"kind":"REWORK","_number":1,"created":"2024-02-01 22:23:09.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/907502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/907502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/907502/1"}}},"commit":{"parents":[{"commit":"76fe1612a082fed41cda56072b570ba98d6b2c83","subject":"Add is-eom option to list-deliverables command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76fe1612a082fed41cda56072b570ba98d6b2c83"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-02-01 22:19:21.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-02-01 22:19:21.000000000","tz":60},"subject":"DNM: list_eom_stale_branches.sh","message":"DNM: list_eom_stale_branches.sh\n\nthis is a blind copy of the list_eol_stale_branches.sh script to\nmake it possible to delete stable/yoga branches where there are\nyoga-eom tag already.\n\nTODO: refactor the two script to have at least a common base or a\n      single script.\n\nChange-Id: Iaaa9702c5b304152492ed2afc56ac07fb9c9fb98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/09ef3cee3767b24eeab5593d74cf3b3a89210a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/09ef3cee3767b24eeab5593d74cf3b3a89210a3b"}]},"branch":"refs/heads/master"},"b6414ac6b54b4ddad489f7566f4b2e3068a4ea8f":{"kind":"REWORK","_number":2,"created":"2024-02-07 15:00:58.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/907502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/907502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/907502/2"}}},"commit":{"parents":[{"commit":"b2daf352f01182e5e63f814c8c52f2dc638c0d56","subject":"Add is-eom option to list-deliverables command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b2daf352f01182e5e63f814c8c52f2dc638c0d56"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-02-01 22:19:21.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-02-07 15:00:38.000000000","tz":60},"subject":"Add list_eom_stale_branches.sh","message":"Add list_eom_stale_branches.sh\n\nThis is based on the list_eol_stale_branches.sh script to make it\npossible to delete stable/\u003cseries\u003e branches where there are\n\u003cseries\u003e-eom tag already.\n\nChange-Id: Iaaa9702c5b304152492ed2afc56ac07fb9c9fb98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6414ac6b54b4ddad489f7566f4b2e3068a4ea8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6414ac6b54b4ddad489f7566f4b2e3068a4ea8f"}]},"branch":"refs/heads/master"},"3c2926ea6f4c8233c082a23b87c749ae42091bb5":{"kind":"REWORK","_number":3,"created":"2024-02-07 16:41:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/907502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/907502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/907502/3"}}},"commit":{"parents":[{"commit":"d2f776462bd481c9c37fd11ca6f487e24a33e3d0","subject":"Add is-eom option to list-deliverables command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d2f776462bd481c9c37fd11ca6f487e24a33e3d0"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-02-01 22:19:21.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-02-07 16:41:12.000000000","tz":60},"subject":"Add list_eom_stale_branches.sh","message":"Add list_eom_stale_branches.sh\n\nThis is based on the list_eol_stale_branches.sh script to make it\npossible to delete stable/\u003cseries\u003e branches where there are\n\u003cseries\u003e-eom tag already.\n\nChange-Id: Iaaa9702c5b304152492ed2afc56ac07fb9c9fb98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c2926ea6f4c8233c082a23b87c749ae42091bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c2926ea6f4c8233c082a23b87c749ae42091bb5"}]},"branch":"refs/heads/master"},"97e76562e055747edb3cf729ec9c613fa3f9b678":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-12 13:19:00.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/907502/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/907502/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/907502/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/907502/4"}}},"commit":{"parents":[{"commit":"88f298c6bc003a50bea67addc46c5de2eecfabc1","subject":"Add is-eom option to list-deliverables command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/88f298c6bc003a50bea67addc46c5de2eecfabc1"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-02-01 22:19:21.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-02-12 13:19:00.000000000","tz":0},"subject":"Add list_eom_stale_branches.sh","message":"Add list_eom_stale_branches.sh\n\nThis is based on the list_eol_stale_branches.sh script to make it\npossible to delete stable/\u003cseries\u003e branches where there are\n\u003cseries\u003e-eom tag already.\n\nChange-Id: Iaaa9702c5b304152492ed2afc56ac07fb9c9fb98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97e76562e055747edb3cf729ec9c613fa3f9b678"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97e76562e055747edb3cf729ec9c613fa3f9b678"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
